)]}'
{"id":"openstack%2Foctavia~762925","triplet_id":"openstack%2Foctavia~master~Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c","subject":"Make focal the Ubuntu default for diskimage-create","status":"MERGED","created":"2020-11-16 21:37:06.000000000","updated":"2021-03-29 10:22:59.000000000","submitted":"2021-03-29 10:21:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762925","meta_rev_id":"f188c635c4e519c19f18654d68377ffd92552cfa","_number":762925,"virtual_id_number":762925,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 10:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-03-19 09:59:32.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-18 21:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 08:01:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 08:01:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2021-03-18 21:56:48.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 00:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 01:37:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-19 09:59:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-28 17:31:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"11534218b5e12936452fc5560dda744fe6769fbc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-16 21:37:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6437a4757186f2d35a99081efdc9a5ef52bfd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 00:12:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adedc97e7750444bb5b90bfc7b4988e1 : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5c2351240444b50b6f1d4600cdab9ce : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff3aa62cfc847ebae5f5d77f4d42475 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bfd98d7b9d94a9fb59c29bfe0bcb18e : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13b77ff4cca44e05b920ef6b98a8ab9b : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c86b20aba1a411083b460f476bb7e8c : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2140110e8b4cc29911830c1e9c2a11 : SUCCESS in 20m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c823587a2124555a89db5be07b2af3c : SUCCESS in 4m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/678faa9ae6db4de2a7942de5f6680f35 : SUCCESS in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/884db395c814486e8665551f8106229e : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b80f7442e4ee43438f491fea1d41f8e1 : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef81935f76fb47078c1364ddc51d879d : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9766f9f558ea4aca801824dfd203a9a0 : SUCCESS in 1h 39m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e8345e374a0440f87b4fc9a78b586bc : TIMED_OUT in 2h 32m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f42b0baf99824944953f354294f5592d : SUCCESS in 43m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b9e7a25d3d13475d849cebe76da6a05c : SUCCESS in 45m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fc8ced8bc1649608cc5947bcdd0569f : SUCCESS in 40m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f87e65f3161d45fcbea31bf865542e11 : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60828856f8784f89b52db84a9366a10d : SUCCESS in 1h 52m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c533cd30c9547d9acee66ddccf86290 : FAILURE in 1h 56m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce85fdfc9a7b4ccf95778400aa9f0b07 : SUCCESS in 2h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ac92fe84c7444f3fa3dea8d2ba2c85fd : SUCCESS in 1h 55m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b89bc2b819234078ae7e6222496c4977 : SUCCESS in 1h 39m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/73201a43921645b480bee40291c8537d : FAILURE in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ff883df46b4765a1f638343a3a38d0f5d55dc0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-15 21:41:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5177957ec2d462e47e9beb6a438818e5d27d7b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 23:44:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c15321271bb460b9517a369f8648baa : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01978e216154c639a4ee4e263f79d4b : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96414c8e27f42338ce8f204f270ae6e : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91733fb7729d4eaaa8314c521104dc32 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc33356d78b24932863a4874f24e2fb9 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c53a43e5554c82b9017e0e1d272e36 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4090a3d3ef461d8c2db1fd8af9ff13 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e0b1de853bf45bd893b7cc44bc64dda : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15aa9e3e6ba845b3a70497779747933f : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/da21b4210c354d3b9db3ca1fd15badb9 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d6542e3091146cda2e57aeee078b1d7 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80ee1f6fe05f443da89f5d31c9734560 : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a160c3f6a84483b96038206873a550e : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8380f227f8aa471681470a98e57cc40b : SUCCESS in 1h 40m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50e3492d16cd45fbbaf054e48b875607 : SUCCESS in 58m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/51139cfa70d04356bb134a3cab10f7f7 : SUCCESS in 41m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77908c20977646d294eb3a1193fde662 : SUCCESS in 59m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8087b0863884511b1038abd00e99522 : SUCCESS in 59m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2a607d2203194139a56effe601cd3568 : FAILURE in 1h 57m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60a5e131853b48cdaf4394c53888daee : FAILURE in 48m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bbef8ca7578b4d45a0d3e74553f3959c : FAILURE in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b34f148e7f345df825a3e9db5894549 : SUCCESS in 1h 31m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/67f5c655e2dc41948aabef95a69d57eb : SUCCESS in 1h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/656b762a7af84530a151231880ed70e5 : FAILURE in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5c0d16f675a7d9286f8df8c1bb36ed71d3a215b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 21:56:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad165c7336c6e2bb393ac257c255f67e1574da83","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-19 09:59:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1471c5c9de80ace87d81096094a232e734b89405","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-29 08:01:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9f613a8f74381ecca0577c88bea471a578e225d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 08:02:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b23d7dc9dc995a1602a1ca7dc0d8a6724f9c0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:21:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7ebdbfd2e5540ecaf6804aef18eba2d : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c1dd171a2c4151b7a5d1952b111d38 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5439da49344549f59028702f654e0d5a : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6799dcc6f0c4a0d99d0201e755e84e2 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bb73a80eaf4ab9854d17152c55215d : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e031be3ecd524519a10d1918d6ecbb6f : SUCCESS in 5m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/882f2cd37e4c4de2b7ad0da1ca034b7d : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d329665c7e5476ba05f1ae7fccccac1 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eeedb4af82d849e2b6e5d5bc1457a2d2 : SUCCESS in 1h 29m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd2f32a2bc834fb997aa881095bd55e2 : SUCCESS in 2h 08m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d8ab3cdbb90496bb9e07b930a69250d : SUCCESS in 1h 19m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f4eaf0dd483747729c765ba71274396a : SUCCESS in 43m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c2b901657f84097ac30816b5da582d8 : SUCCESS in 35m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6121f21e7280402f8e3467e68d3199d8 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d77d1a73c0b027ca93457acb03a2afe74e4553b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:21:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f188c635c4e519c19f18654d68377ffd92552cfa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:22:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/211af645462746e9a0f7d09fb609bbd9 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e442b0c11f99462284fa9a45c1766e41 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40d011ab7f1f7141041440b760a8e68671979016","revisions":{"57c211f9db413d0588d431a6c381a753dc5d9831":{"kind":"REWORK","_number":1,"created":"2020-11-16 21:37:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/762925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/762925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/762925/1"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-16 21:35:41.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-16 21:37:05.000000000","tz":-480},"subject":"Make focal the Ubuntu default for diskimage-create","message":"Make focal the Ubuntu default for diskimage-create\n\nThe current PTI version for Ubuntu is focal. This patch updates\nthe diskimage-create.sh default to focal from bionic.\n\nChange-Id: Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57c211f9db413d0588d431a6c381a753dc5d9831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57c211f9db413d0588d431a6c381a753dc5d9831"}]},"branch":"refs/heads/master"},"40d011ab7f1f7141041440b760a8e68671979016":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-15 21:41:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/762925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/762925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/762925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/762925/2"}}},"commit":{"parents":[{"commit":"71c7402bc964ebac8361b7522ae98c58c171eccb","subject":"Merge \"Fix provider driver quota handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71c7402bc964ebac8361b7522ae98c58c171eccb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-16 21:35:41.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-15 21:41:24.000000000","tz":0},"subject":"Make focal the Ubuntu default for diskimage-create","message":"Make focal the Ubuntu default for diskimage-create\n\nThe current PTI version for Ubuntu is focal. This patch updates\nthe diskimage-create.sh default to focal from bionic.\n\nChange-Id: Ia380b712d9d40d702b99bb3b04e1f45dcb60b96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40d011ab7f1f7141041440b760a8e68671979016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40d011ab7f1f7141041440b760a8e68671979016"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
