)]}'
{"id":"openstack%2Foctavia~762927","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","project":"openstack/octavia","branch":"stable/ussuri","topic":"fix-gate-stable/ussuri","hashtags":[],"change_id":"I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","subject":"Fix pep8 failures in gate","status":"MERGED","created":"2020-11-16 21:48:03.000000000","updated":"2020-11-17 13:06:55.000000000","submitted":"2020-11-17 13:01:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762927-1605618091595-b2dd4156","meta_rev_id":"1f1640137c5e85d844ec878d631094a5e2ea0f0e","_number":762927,"virtual_id_number":762927,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 13:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-11-17 07:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-11-17 11:17:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-11-17 11:17:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-11-17 01:29:29.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 01:29:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-17 07:27:19.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-17 11:17:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-17 13:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ecbba4cef4d29fc0585bd244b772a9ceb3e1c67","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 21:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60c895817fb71b08dfe791bbf4cb831989652ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 00:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73667635a64c4ed88a0ee4b11f913ab4 : SUCCESS in 8m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef43b9e86fc147cabf3a9b3d6d369c1f : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6e093e9e36423b9c919f2fc9713c04 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc19dadeb2a6478099a2734e9537d5d1 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9500e452b9b641b1b2b437d3c736e8d5 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a36befdc3549069931bb31dcbe22aa : SUCCESS in 9m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a3fe0863bd464bc080f963509b03f11f : SUCCESS in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4849beef903443be9a7141a727a56734 : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0279987a85284361a04ea21b85551026 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e8d5f85e41c42cb84b102506a360266 : SUCCESS in 1h 21m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1061b5757e1e43fc900eeb08fcd860c5 : SUCCESS in 2h 02m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e2365aecdb9479f8e0e07503b82563b : SUCCESS in 48m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cded0e4351af4c9884eed9e9975182bd : SUCCESS in 38m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4ed15dc17284548bc0c82aca05fc5b8 : SUCCESS in 42m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b6a18dba547464895b7630c3b276a3a : SUCCESS in 1h 18m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1814f3dbaa4246eb9159238900c42a0b : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/85fbd26890d14c14b57bc0c727b66802 : FAILURE in 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/423962a73cc74e6b8f465dbacf7f39d4 : SUCCESS in 1h 50m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3e830e7b4d8805ce0bb9f9bbcbff0d7514a1a72","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 07:27:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"637f504f6c43aa24d5a4cb2a74af3819883127d9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-17 11:17:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis is fine but I\u0027d have preferred to set an upper constraint version instead. Future lib versions may break pep8 again.","accounts_in_message":[],"_revision_number":1},{"id":"cdaafaa4ef050e078d857e90dc6ed3023916864e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 11:18:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c3fbf50c5cf5a1cfedd57c52c72b42eab1fe319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:01:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f22a923e4e4c472e8a96ebd07b814fe4 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a79f6f40bd45e69a9c140b1417b12d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07114741606a4f968117260f3cfb1b69 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/367baadce81c4d419c7bb1f709e3055e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de2ec33577140899c0670a6f5a0e147 : SUCCESS in 18m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d780886574154080b875f9b656aab507 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e1ec533c5c045358bc8b486e627ec57 : SUCCESS in 1h 12m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb6b4836b62748a2adee8a112d0e5872 : SUCCESS in 1h 40m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c76d8e11b1f149bcb36cde64a688f47f : SUCCESS in 55m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e7648697174b4c1694159e1cc3ce6085 : SUCCESS in 1h 20m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84ba57f92d764d19920a85a82002d1b7 : SUCCESS in 54m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/106d0086c3224f53b17752745575c33d : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b20095d9c16280d4b9a23fde0abac863ab9e9ae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:01:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb05fe3eb7397c98645a32e6f86c716192187f7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:06:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dce2668deb4e87ae85f734b93d0f55 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71c214c8f879260980a4688d77d21e98353f4da1","revisions":{"71c214c8f879260980a4688d77d21e98353f4da1":{"kind":"REWORK","_number":1,"created":"2020-11-16 21:48:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/27/762927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/762927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/762927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/762927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/762927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/762927/1"}}},"commit":{"parents":[{"commit":"c366169a489e9fb53bf2092d19fc1d35200562ca","subject":"Merge \"Update devstack plugin\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c366169a489e9fb53bf2092d19fc1d35200562ca"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-16 21:44:02.000000000","tz":-300},"subject":"Fix pep8 failures in gate","message":"Fix pep8 failures in gate\n\nSome package must have moved forward recently, causing\nthe pep8 job to start failing.  Fix the issues since\nthey\u0027re all valid.\n\nAlso changed other messages using an exception string\nto use str(e) to avoid any possible py3 issues.\n\nFixes the grenade job to properly build the test server.\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n\nConflicts:\n  octavia/amphorae/backends/health_daemon/health_daemon.py\n  octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  octavia/amphorae/drivers/health/heartbeat_udp.py\n  octavia/controller/worker/v1/controller_worker.py\n  octavia/controller/worker/v2/controller_worker.py\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n(cherry picked from commit 57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71c214c8f879260980a4688d77d21e98353f4da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71c214c8f879260980a4688d77d21e98353f4da1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
