)]}'
{"id":"openstack%2Foctavia~762928","triplet_id":"openstack%2Foctavia~master~I3c830a9f0fd21ca10238946721c0f75d01d3ddff","project":"openstack/octavia","branch":"master","topic":"redhat-cloud-utils-growpart","hashtags":[],"change_id":"I3c830a9f0fd21ca10238946721c0f75d01d3ddff","subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","status":"MERGED","created":"2020-11-16 21:51:13.000000000","updated":"2020-11-27 11:27:40.000000000","submitted":"2020-11-27 11:25:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762928-redhat-cloud-utils-growpart","meta_rev_id":"07836fb0ac60675db0464f7f9a8c6c68617f53d6","_number":762928,"virtual_id_number":762928,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 11:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-11-27 07:55:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-24 08:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-11-19 22:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-11-27 07:55:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-11-27 07:55:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 14:00:03.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-11-19 22:53:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-20 00:43:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-27 07:55:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"b9ddcdb5402adb11f3d0c91739e5bccf64d24636","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-16 21:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a26745845c954e524498cdb75545ac366a9929f8","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-16 21:52:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b1a0a0c0eae0aca9bb2133a7c52d4ca6ca201ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 00:42:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bcf0ec7c51b4e42a1c4e4d2d52e9321 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b650444cbf14d32bc7bdcc5789a6be2 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65486c0c5e67498c8e73b1878b77bad2 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02459d155ccf4b39bf36b21032395897 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f30593173a3435b9e5ff07a559994d5 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2206cb2ae2974c309e9fc36514c8efbd : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e593dfc1ed4803a2be43ce80cc752c : SUCCESS in 17m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85e461b47bcd40f5a19b2ca07a146d52 : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6bc15540fec342e9ba4d52270990c654 : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2bcfa1ff37f34b5ca80103231c9249b6 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9cc751fd0d64a8eb48f9d08164e0174 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f7b02bf7a884484a4b59b1fc86e52cb : SUCCESS in 1h 18m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfb9574b8b5b4a16a0bdd69ab85c5e6f : SUCCESS in 1h 49m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19c69f98c43d4f57a9c4d330d6b0f6aa : SUCCESS in 44m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/75e1519ab121419486a6d2ac85b27974 : SUCCESS in 40m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdac0e124fe142beaa65996c25faed48 : SUCCESS in 1h 30m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc7a389a4adf49f8b0a26cbaba28b299 : SUCCESS in 50m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7637428468943c189717d2485bed7c0 : SUCCESS in 1h 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fc9b8506d8440da8771558fa6ea7f75 : FAILURE in 2h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6480e3b5870a49cc994bbe36357bb57c : SUCCESS in 1h 54m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92eb790623ed4d85a0b714f7e1d1de36 : SUCCESS in 1h 46m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/808ba37315a84f8daa5b1bfd950b56c5 : SUCCESS in 1h 45m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b79b3c889cba4a70acc127abad16eaa2 : FAILURE in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93c5d4a5c4d529855a61868edee170343140f002","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-19 14:00:03.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\ngood catch, I wonder if it needs a release note.","accounts_in_message":[],"_revision_number":2},{"id":"d18ce5464db9ce2c085716510701006840ad0326","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-19 22:23:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1bde4f6c5b3c252f5e7f2ff66cd78ef66a95e4b4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-19 22:53:40.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"6ae43bad39af065bb89e1b3ab4de2916adc0974e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:43:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d51802f8cb40f78a868bcbd7420e23 : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e19b7fb2463f4c1c8624c84c76451eb2 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b3b85fadd94db2b4159894906715a1 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/159b7bc8a0674cf9bbdbc46b0b4463d5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c4cb9f2afd4491c8509ae8e07dce7c9 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9343993d04e040be8ba027d73fe2de60 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438cc08e160348b68e6aa170b612e66d : SUCCESS in 18m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502aca4d60a84cc0b4f89595f1f36c56 : SUCCESS in 6m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/16b43304f3aa49ce8c604430141b0410 : SUCCESS in 7m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/88e6ca43a8344157b6ccd5e5da3367f2 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0d6d5a65a5b4c2490e01be193c00ec3 : SUCCESS in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0df75e466e5b4cb288ed3a290210defd : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2092819c3aaf41d18ee072d132f7f530 : SUCCESS in 1h 33m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f5551c60f674462a292b55cffd38c5c : SUCCESS in 1h 46m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70b992a7ce5a4565a9f7c7892ad5b906 : SUCCESS in 39m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4113c578e461452bb4a6e934ed42a0f5 : SUCCESS in 44m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1c0f3e1e5bf444ba0575d3b79477790 : SUCCESS in 45m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/962c6c8e1967471dba99709cbaa015d3 : SUCCESS in 1h 00m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/673799a0e2d747f19c02cca024442983 : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e0fa803c8a646c8bf19ac640f7b5d9f : FAILURE in 1h 58m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4897dcf13f314086851c52cd97b18c5a : SUCCESS in 2h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/108b1294aa6248ad9255c2ba36e547e3 : SUCCESS in 1h 31m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2045a37091804042b5ba9ccc294f47f9 : SUCCESS in 1h 43m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/90f97a56eea648eabd929e9d8542d412 : FAILURE in 54m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c43fc3e9a0715630879c4d2b79bc413ad319abd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-24 08:38:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"90bb616b7cba8afcf2831c3036c6327ca6ac4929","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-27 07:55:44.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce31e21a573bd383d904914e56557a23993a6565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:55:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95b340be77c14fcce438097be3e09aa071d6f51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:25:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e84e2331dd5b49039073eba586f91e81 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10444599e41744f988ec280d7223f393 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff9c4ad1011b477a9cd7886196f9a125 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7648d7fdb10f498580693ea851bb82f1 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86f7942fc444c95af8c3325046b0ea4 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b6dcd796f049f5b09b4c43c8235203 : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ecb2a2bb9813450995e113678e5d4c13 : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9f2a7d856d94b19bcc52559700d5d12 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0fe7585613dc48eeacb23001be6cfe5b : SUCCESS in 1h 18m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/509ff6346b714a229275f53588adc7b1 : SUCCESS in 2h 25m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b43a54bc4164eaf8a2d99651f212c97 : SUCCESS in 50m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/342a4e2fca5f414ba41f9f0406c1c0bf : SUCCESS in 39m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b269203c1824112a24d2b85bf8e3e30 : SUCCESS in 55m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14d57262d19242fba6e9a3d057fa7f62 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4100548c1c9cda2be279755803c77244fd38c3f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:25:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"07836fb0ac60675db0464f7f9a8c6c68617f53d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:27:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47195d8312a8470d85f0e9367ec6efa7 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/920d9157687840eaafbb4539459feb5c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d59fec8a0c8011949c547c4f49195c41b9d21a5","revisions":{"7655763b1bccba9336926b0b3ab87118bdd44a93":{"kind":"REWORK","_number":1,"created":"2020-11-16 21:51:13.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/28/762928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/762928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/762928/1"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 21:45:07.000000000","tz":-300},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based distros\nwhich means the root filesystems are not grown on instance startup which\nmeans not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7655763b1bccba9336926b0b3ab87118bdd44a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7655763b1bccba9336926b0b3ab87118bdd44a93"}]},"branch":"refs/heads/master"},"436ea087bfa4c217b0d195c4172ab904da974c14":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-16 21:52:28.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/28/762928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/762928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/762928/2"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 21:51:33.000000000","tz":-300},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based image\nbuilds which means the root filesystems are not grown on instance\nstartup which means not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/436ea087bfa4c217b0d195c4172ab904da974c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/436ea087bfa4c217b0d195c4172ab904da974c14"}]},"branch":"refs/heads/master"},"6d59fec8a0c8011949c547c4f49195c41b9d21a5":{"kind":"REWORK","_number":3,"created":"2020-11-19 22:23:09.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/28/762928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/762928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/762928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/762928/3"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-19 16:49:52.000000000","tz":-300},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based image\nbuilds which means the root filesystems are not grown on instance\nstartup which means not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d59fec8a0c8011949c547c4f49195c41b9d21a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d59fec8a0c8011949c547c4f49195c41b9d21a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
