)]}'
{"id":"openstack%2Foctavia~763092","triplet_id":"openstack%2Foctavia~master~I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01","subject":"Add nightly test image build job for focal","status":"MERGED","created":"2020-11-17 22:05:19.000000000","updated":"2021-05-05 19:53:49.000000000","submitted":"2021-05-05 19:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"763092","meta_rev_id":"10cc40b8fab59b2f9fb28a36e9e0543aaf90cdc2","_number":763092,"virtual_id_number":763092,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 19:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-04-28 16:15:51.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-05 16:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-05 16:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-04-28 16:15:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 00:38:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 01:27:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-19 09:58:40.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-05-03 09:39:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"781d8e1e9fed63c0dcbdec063bba1baf44adf6cc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-17 22:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb3ed53112ca6f1ae4628553e68c257a2c361f02","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-17 22:05:36.000000000","message":"Patch Set 1: Code-Review-1\n\nTesting in experimental","accounts_in_message":[],"_revision_number":1},{"id":"6be3c47074d46e79c0ee910ab34dceed9fb7f27e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-17 22:05:53.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"5126b4bd56bf4988805ef2eb8970bcb9db30e624","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 22:06:12.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Pre-review pipeline experimental does not allow post-review job publish-openstack-octavia-amphora-image-bionic","accounts_in_message":[],"_revision_number":1},{"id":"d847019d2427be07d95b08fa8dd8ff03c55af0ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-17 22:07:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b907b487398d897b0593da9de1232ab58048d84e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-17 22:08:54.000000000","message":"Patch Set 2:\n\nHere is a reference to the failing runs: https://zuul.openstack.org/build/aac32f9cdb1c4583a3b53425f22bb6dd","accounts_in_message":[],"_revision_number":2},{"id":"b65e4239f0e7531c3694bc82b7f434dace975629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 00:38:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4a3a994ec7e4d53837a825c35b40b20 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/939703e91e8141528938a2b5eb1c8ebc : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2162c652a59e4d508600bda2ce0f132e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/442c4312a0f84fbaa035430e9445da0b : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e07eeaace7462c9ae5b3ddb9c019f6 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92912a01870d4682a0136538b6acd389 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af1e00c66f24d71a219d7f60c003a29 : SUCCESS in 18m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a098b6a92484402699d52fd4d7b36a24 : SUCCESS in 4m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d9e98e9f61c342828d95af35cdf167bf : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67abb4af11594af988c139b94e82f4bd : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/068fb09233204ab390e5d182be7bf4bc : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6d6880de0b10405a9815254e387bb061 : SUCCESS in 1h 26m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc54132c34f7470aa192f76586b2c42b : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50ac2d116eaf4a7b9da0b3cf9721e9a0 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7af9316964ed4d85a6d6f5d5c9291082 : SUCCESS in 36m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cca257370df455fa493d437a1e8ad37 : SUCCESS in 51m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e26096b7cb2d4ed29b03fe1bc5e504e5 : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a22722638a9945f9a404cfc0e950ac30 : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c43ac534b02940f4bec6f360b49726ef : FAILURE in 2h 09m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d984ee9cc11743cab8094f1d40ea81cb : SUCCESS in 1h 55m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/75a1525d942f482195680f6f58d1ca7a : FAILURE in 1h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5b1aa85bd8a04a799b7c8bc05b64040f : SUCCESS in 2h 28m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cab020bc7fa54691be5d3ece1dad7054 : FAILURE in 42m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f4d935d3628ce9cfa3bbbabef442511100d11ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 01:27:48.000000000","message":"Patch Set 2: Code-Review+1\n\ncheck periodic\n\nDoes this work?\n\nAnd I was just updating the grafana dashboard so if this merges I\u0027ll add it to that.","accounts_in_message":[],"_revision_number":2},{"id":"10635b229260de676b9230adaa5151b38cb0a0a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-15 21:53:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a188b634a59422b1e3d19d891c2623b825cd34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 23:43:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c930b849d5ed4ae185695e151b38d1f9 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08d7a20d9b2a48c491a6e082f58a0fdd : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f07ed6305641808718c3e59bff445e : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2019b5a54ce4caba932f07966de7479 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81349b0419c04a488bafe52ba268b632 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc1a9205b8004d939a34e5544b57fd3b : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4246289d9124df48b215d6c888b9bbe : SUCCESS in 18m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e6d802922e04dfeb5c8d7b1c2b9bf00 : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/16161f8826da4e35b344921f583aed11 : SUCCESS in 8m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/416b2806ec5b45a59fff48a0e7e259d3 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb9e5fced60147039f29aa29002d3f12 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7c0c1092050417aa5936434f77e799a : SUCCESS in 1h 28m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dd0f153ef4c45a9a3b80252b9b37fa0 : SUCCESS in 1h 26m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82c1e383c48741aa8c49cc40113557d5 : SUCCESS in 1h 09m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ca8c27232faf4f77a8721c3ba2a4dccc : SUCCESS in 31m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2397515358224edba939c8353185e99b : SUCCESS in 41m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa02c47e6c8b42b999cfb29f9e1a52d0 : SUCCESS in 56m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3370a8bd8f8846f188d867d63dec666d : SUCCESS in 1h 45m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6855c137a42148a48026f1451783fdf3 : FAILURE in 48m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7f0cfb8c674932b1e183c21286a2a5 : SUCCESS in 1h 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a4d589fae20492d9f4b2a0c13a97a32 : SUCCESS in 1h 46m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b922dc121ad46fa9fd053b1f1b75d7b : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/aa0bb804d81846bb8b60f47a452ddce8 : FAILURE in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"35bdc90d39069776288c577ec04be7c7d1f6e7ac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 21:57:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"44fc4bc5c784e924dcd14d6a42c84da5a201662b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-19 09:58:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c520e96448c50783ea8b50db1aaf538019403227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:38:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f876b2cb10cd9472b3a59b64b5f518c84088aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 23:23:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6a258facbe0409e99e962a790f35c8a : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1060dd9a8a1e4405bcbf4c8c6f97b83b : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4a4373306c468da6b775de241ec3f8 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df4cd3a74ab4d7a94a5e46890952fad : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21038b0aa56f452999bcdaf336c808b2 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c1c5e5fdba4167b8ead073ae80af07 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0486dfbfc34518a315711d3f12eb42 : SUCCESS in 12m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4f4a7852c68e4548bfa86e6c1dc35a6e : SUCCESS in 6m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/68acedb1bfee478d9f9f709a418cb5b3 : SUCCESS in 9m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb6e8ea57ec2428c924f96d223ddcfba : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dad511961f4c47429c46aa4968cbd4f4 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7e125cccb824314b10928c21eb6719e : SUCCESS in 1h 26m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbab70f949844bc5acfef378964872a2 : SUCCESS in 1h 41m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2cd6dc4d4ab04d379714b06cd65ee62d : SUCCESS in 45m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/923dc65bf1d3410fa7810f097d3fcc4e : SUCCESS in 38m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e55072b142cf4a22aaf5b7fadf98d903 : SUCCESS in 42m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a3f8f82732743b5bfa21338c931b085 : SUCCESS in 1h 07m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f915203f62414f01b4f78b89b4114bdc : SUCCESS in 1h 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb77253baff343a79a12d77f51633aec : FAILURE in 50m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c8aaf0600ebf4f46b1d0e3efe55cf1c1 : SUCCESS in 2h 03m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ebfe99258d4243efb3f84fa8b89b36b5 : SUCCESS in 1h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cfa043db8cfd4b67a13812d961950c1a : SUCCESS in 2h 00m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9f2fadb272894af6b6d736cfe9a27904 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3be5c6380ef8dc2872cd59a68c9edf87c684c884","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-04-28 16:15:51.000000000","message":"Patch Set 4: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2cb912fa2823d80b82c48196c32b69a73d01dd6b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-03 09:39:42.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2fa76a4ba5aaaaf89b6ef3a71afa34c06d95131","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-04 22:11:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"14825eaf715794a29be8959c82a8b176b6a9d4de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-05-05 16:15:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"175072f29da6288d20a04ab8113e597c7ede834d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:16:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cc902a9c0fa030ba6695c8e1211445198f5b1fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:31:02.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab434aa0fc2744b3b977a38c8e22aa9d : FAILURE in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bbf112d4654ab0a283d563b044224b : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aba4fc1110649aa91bf5bbcc7aaf8cd : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f50297a94114124801ccb6156dd275f : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6085013f23a4d69b8a623f3f3365c1f : SUCCESS in 10m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/843ede06ea32468f916cb763f912ff57 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f943fdda9904ad58dbf2e57ad77067c : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00edd1a8cc6c44ccabf422e6c9480908 : CANCELED in 12m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ffca9571e3c4f71987fd132d3b7eb5d : CANCELED in 12m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/98e68d26b8744db5a0f63b67d314cbcf : CANCELED in 12m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcc4e049c96940bbb7e969ec65e06ed5 : CANCELED in 12m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/239bad73d3064a0b9f10e4998cf88bd2 : CANCELED in 13m 09s","accounts_in_message":[],"_revision_number":4},{"id":"780edcf4245425a498ef5147976a610065c648d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-05-05 16:37:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"edbf90d5003298a2cef260d038d19b4a0d856f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 18:25:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb97fff7688848b59b04f95d407c4856 : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f4b6033836340bdac3484df59216797 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994df01a222a4af9b03343ff8c9c1ccc : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c55b4ba54d624ca3854588854afd86a1 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ea99380367432e9bed8264a2552ac5 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57754d52e3094b0c8d2eed7b3404ffd4 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19da5855921347459f11cc7f0c805a42 : SUCCESS in 12m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/38e28d4e463c494cbbe41b72fd0a7124 : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9006428f63184009980cebbb4173e3e2 : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/936be160032546e7a7243ac22e8897a8 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/639572037ebd425fbf4e002250c11199 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba80dbef63274efa9385e1fc285d735f : SUCCESS in 1h 41m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b798c26bf9420b9d6d538c73a98daa : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6275a3af95154aababaf55d5c312777a : SUCCESS in 40m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb5e193ac4c94c7eafe1baa69d4c5079 : SUCCESS in 35m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82c8fe391a244bcbbffd415b5c6e916d : SUCCESS in 1h 11m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d0ac2bd1e52547fbae8b04dfa3cff45f : SUCCESS in 1h 45m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4dafce6315a491181c004a2a481ad14 : FAILURE in 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/707918b6b6204a1b9a9911796c08e46a : SUCCESS in 1h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39247da6312b4d7b82d3f41a58dba10a : FAILURE in 1h 37m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b81ded9a0ec742c3b581da71d00ee8ba : SUCCESS in 1h 38m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/67609c338f6f478aa6b3bc7f8dd72911 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c6b20213273585413cdb318ca15e07a76a800e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 18:26:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"63105ef01af06a902b3aa3d8deaa032f536cbff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:51:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/863db23cf8da42c89ce09eeb71c79361 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbcb5057c9f427b92e14ef5035d8299 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcf0a5fddc634778a0841553563407c0 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55fd68cedf1d4d61b86fe111a3d7da90 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de02d36f7a44552aeda32f93b4c45f2 : SUCCESS in 14m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/488dfecd7cd34d90a644084a398899af : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84e852a50ffd49959593177c5e8f2e01 : SUCCESS in 8m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b661da5425c4b3da13029c1771ee3e4 : SUCCESS in 1h 14m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf847b2b2cfe4e9b954cef2c3445d275 : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3361ce5aa08e44de9ed12df9d9d86a07 : SUCCESS in 44m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f736d9c64fda4f8783618470792b525e : SUCCESS in 34m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c77126e4d3834008acbc266f87635023 : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":4},{"id":"164caff28f260f0944bdba4883aa13b685b53f39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"10cc40b8fab59b2f9fb28a36e9e0543aaf90cdc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:53:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65fd74c73992401db75ea6b3ddb841bd : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c1910f37ed2ff11bd9216a01cbb9b79e93f27e45","revisions":{"11fa4831542ccc43444fb1383541f90973de08c6":{"kind":"REWORK","_number":1,"created":"2020-11-17 22:05:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/763092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/763092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/763092/1"}}},"commit":{"parents":[{"commit":"f96b0986cda925d471073620047346c70b73927a","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96b0986cda925d471073620047346c70b73927a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:04:06.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:05:18.000000000","tz":-480},"subject":"Fix the nightly image publish jobs and add focal","message":"Fix the nightly image publish jobs and add focal\n\nThis patch fixes the nightly image build jobs that have started failing\ndue to the parent nodeset moving to focal.\nIt also adds a nightly image build for a focal based image.\n\nChange-Id: I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11fa4831542ccc43444fb1383541f90973de08c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11fa4831542ccc43444fb1383541f90973de08c6"}]},"branch":"refs/heads/master"},"aef84dd28db8bdaa47ca0a4997665fb058176264":{"kind":"REWORK","_number":2,"created":"2020-11-17 22:07:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/763092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/763092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/763092/2"}}},"commit":{"parents":[{"commit":"f96b0986cda925d471073620047346c70b73927a","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96b0986cda925d471073620047346c70b73927a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:04:06.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:07:19.000000000","tz":-480},"subject":"Fix the nightly image publish jobs and add focal","message":"Fix the nightly image publish jobs and add focal\n\nThis patch fixes the nightly image build jobs that have started failing\ndue to the parent nodeset moving to focal.\nIt also adds a nightly image build for a focal based image.\n\nChange-Id: I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aef84dd28db8bdaa47ca0a4997665fb058176264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aef84dd28db8bdaa47ca0a4997665fb058176264"}]},"branch":"refs/heads/master"},"a39ec49e9b3dc20447e3714d18f24c77a77befb4":{"kind":"REWORK","_number":3,"created":"2021-03-15 21:53:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/763092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/763092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/763092/3"}}},"commit":{"parents":[{"commit":"71c7402bc964ebac8361b7522ae98c58c171eccb","subject":"Merge \"Fix provider driver quota handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71c7402bc964ebac8361b7522ae98c58c171eccb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:04:06.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-15 21:53:42.000000000","tz":0},"subject":"Fix the nightly image publish jobs and add focal","message":"Fix the nightly image publish jobs and add focal\n\nThis patch fixes the nightly image build jobs that have started failing\ndue to the parent nodeset moving to focal.\nIt also adds a nightly image build for a focal based image.\n\nChange-Id: I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a39ec49e9b3dc20447e3714d18f24c77a77befb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a39ec49e9b3dc20447e3714d18f24c77a77befb4"}]},"branch":"refs/heads/master"},"c1910f37ed2ff11bd9216a01cbb9b79e93f27e45":{"kind":"REWORK","_number":4,"created":"2021-04-23 20:38:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/763092/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/763092/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/763092/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/763092/4"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-17 22:04:06.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 20:38:21.000000000","tz":0},"subject":"Add nightly test image build job for focal","message":"Add nightly test image build job for focal\n\nAdds a nightly image build for a focal based image.\n\nChange-Id: I03ed95e876c0fb8dc9ac44986d42eb664fbcbb01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1910f37ed2ff11bd9216a01cbb9b79e93f27e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1910f37ed2ff11bd9216a01cbb9b79e93f27e45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
