)]}'
{"id":"openstack%2Foctavia~763249","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6693be3e2d2f0e34f19f07f55b0429f8fda317b2","project":"openstack/octavia","branch":"stable/victoria","topic":"task/41009-stable/victoria","hashtags":[],"change_id":"I6693be3e2d2f0e34f19f07f55b0429f8fda317b2","subject":"Handle listener not found during loadbalancer status update","status":"MERGED","created":"2020-11-18 19:28:14.000000000","updated":"2021-01-14 14:49:09.000000000","submitted":"2021-01-14 14:47:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763249-task/41009-stable/victoria","meta_rev_id":"f7c4472ee6e293e083797f2349c1ed3bc825ae97","_number":763249,"virtual_id_number":763249,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 14:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-14 14:49:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-14 14:49:09.000000000","post_submit":true,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-11 06:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 22:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 14:05:04.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-12-14 09:05:07.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"fc8e7c5f8ce68b87338d383cb7a3ecd3dd4ab8c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 19:28:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d309de99063e3345ddac77453c9ec1b751d750a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 22:11:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12d7903508e54ad4b28626f6385b79c0 : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce25330c906249eab68d08f1023cc848 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2ebe087e024207b1b444d15066d7c3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284841c1a19846359990a16e14109f90 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/646c2a4505af4e4b96f13b31558f477e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e971898f6b45009f4bff263576606d : SUCCESS in 11m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dc64f1245eab49c28b860c093b694b24 : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/19f3e087a92e4905b8b1084fded4f714 : SUCCESS in 9m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/77593735642c47cab0b2ee9fd84a3244 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/471e197ca6604f2ca562dee64d64e05b : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0da448fe66e47b2bf40ad15a94b8844 : SUCCESS in 1h 26m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/235c2c99fb8c4bb0973747f3418584e9 : SUCCESS in 1h 50m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90e9fec1107c4f988480919e821aa489 : SUCCESS in 41m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a4aaf3bd9c454397947fc92e69d5d158 : SUCCESS in 42m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02689ea11554490ca73f4fb85e175dae : SUCCESS in 36m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4008a67eecb9448aa85091f113faa897 : SUCCESS in 59m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d15afe298fb4dc7bebdb3b0b420625d : SUCCESS in 1h 47m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/54f35748384a4e0190d1a13399a51509 : FAILURE in 2h 18m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8208cdee5d9b475a86b8b28644fd0efd : SUCCESS in 1h 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/85e6db235166485e9bbaf031eec8e569 : FAILURE in 2h 16m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6b6c20ac73a54e78bf0e463d1e0e7215 : TIMED_OUT in 2h 34m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51ab8fd711e04e7bb5cdf041bd7a75e2 : SUCCESS in 2h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1025b018d13acd4768cc4cde44f7677724289dff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 14:05:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"335fc64a0adad795893cde2cba7b97b32c1f2e7f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-14 09:05:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7efa48d6427fc7350806e7472dcafba11fa7116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:05:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e520c4b398a0cb48f69afbfe5e47cf07a85324d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:29:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c076046634844138b5fd9eaa60dfd4aa : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9993cfddfdf649afb13a680bfc0375c4 : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f95d4f5ea04a8080eef849a485204e : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac6ae98a318c452e9a07ec9f87fccc2c : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f5d2267b0b4cc891408af4aef655e5 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9ad0937d99d342f3a2d4deab40fdeff3 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd56f25a1a5f45cfa795ecd9b51ab4f4 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5ebe23185354aad9f1286c409088c9e : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61892c1bb66742bdb20be255472210bb : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4049c6ac293c4bb0abe13eef3f4e2fb2 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45136e539f974038ad44e5952bfe4a60 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e63a639f19e247078f96b856d93117ce : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e86032cd220544baae37a019a27740b5 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"4a0331b96e56a15338434754ab95bef35a9eb0f4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-14 09:39:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b64a38e278976f74d23328df65f57cf3d2d456e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:19:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7fde1d5589940b2abf955cd9c673796 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21420b11b2174272bacd99781743d78f : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e04d9980b8c4e8298a3d8c68ba06ac3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea5420326407493f8130d407b274f0b1 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbd2c0c89ae3414a87df88ced288da70 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a61bf19e8c944d0b0a76b89dfebeaa1 : SUCCESS in 11m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ffeace99f87941fe977ed19d0be237b7 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/388284fc356d4b93abc8e1996c6aaf55 : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/541eb5cb540544b7a17891da927cd400 : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0951dcd0f15342faa1884f04e7d2ebfd : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e7e32d5c6c341799ef2fc72e9435900 : SUCCESS in 1h 22m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/120b57e593c64f6d8be14d2f349c14c5 : SUCCESS in 1h 46m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/360af58f8d8f47359cf2accba7a8c658 : SUCCESS in 51m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0bd0605fdbca414a9ee237f9bacc4f2a : SUCCESS in 1h 19m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8ce81c55cf345c4b8f07e2827a1f9bd : FAILURE in 1h 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4218b380a115463ba747233444c89035 : SUCCESS in 1h 13m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05271a90072246b483470a803d1ec9da : SUCCESS in 1h 49m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f04ca400901a4c6d9c42de22cdfb3788 : FAILURE in 2h 04m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d4a2b3f9f9f4403d8aee5230cdd828b5 : SUCCESS in 1h 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d3f655c4583d4263b4e2e836759aa6da : TIMED_OUT in 2h 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6910741a025146d5adb9f83275841690 : SUCCESS in 1h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/af1e744f4b964b3292df4cbfdedf0077 : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18ebdabc060d601954812e73cd93ac9c0ffd36a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-11 06:19:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4a2d32565de21b629042ba15b339fa033a114594","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-11 06:19:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb81bc3bd1f4b20d7481e9fa2c57f9297dfe7b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 08:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0ef4e7efbb640ea94d2bfaa6d89c8aa : SUCCESS in 11m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3e52c0fc64f48ec97df1d6afd514599 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28b6dd2163174afcb42a21a28e5c5e9d : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0548e3e8c1de4da682d83e2746437ff0 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca36d6789f454112bb620a366dbd3da7 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b230e5dc9c75489b9b19155efba3b2e0 : SUCCESS in 15m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93b43879bea24551ac5aa251ddc600e9 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dbb52e9032a44f0a8273d241de030308 : SUCCESS in 11m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cfee4cd7a33d48b1bf9363bcce143f92 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d412f2802464168824a8b4037718a81 : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/830376ac3e7e47ceb9228e18240f438e : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0578a9330c914109a1fee640f1166bef : SUCCESS in 1h 57m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47eff36b98ee452b8adafbe0ccbcf298 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/19664242505943d784b990578599ef19 : SUCCESS in 36m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/910c3459efba46a3ba9318bfd30ad397 : SUCCESS in 49m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97280175db6148238cb5ce6c67e13015 : FAILURE in 28m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/09e090ea3ba34abbb5f7252bc4b08ce2 : SUCCESS in 1h 44m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b9bf7667a6d24bddbfb5322ebf29e397 : FAILURE in 2h 00m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0c7a757d5de148178842fa632e69c57c : SUCCESS in 1h 50m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cbb213ea67dd424d9cc04975a23c1b7d : SUCCESS in 1h 37m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bab372bb2a4d4055bdf401c8595f2b9d : SUCCESS in 1h 55m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b4061088590e4efda6e44aa646543b44 : FAILURE in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4299d0db274d2001abb7959c3e65d5217fea058e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 15:49:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"43f73395286eb4dc044bc98bdff72948be93a7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 18:18:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c657910f17984cd28cf90142e4c7198f : SUCCESS in 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dda736d3cc24170ba59b84d86157aaf : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad4b14e5de747aaa9bf89e49115256c : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e88f505573e4a6faae019b9aba3d6ec : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f14bb2091249a7a60a81e4071783ad : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03888339def54adfa11d93980374e606 : SUCCESS in 13m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3bad610fe3aa40d2af115e2a3c433013 : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1bf89ca166314537934ccabd78af5cae : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4dd22235c0b54ac399869244ddbea94f : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd6d0c75e0284b45b9db191be77c02f7 : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba5e61d59b7e43cb80a35fe2be45d213 : SUCCESS in 1h 24m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/823028e5afba455a821fe53b8b1b70df : SUCCESS in 1h 56m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef596f7815094d1dacabd164b0a6ed5f : SUCCESS in 42m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/53b351fed17c471f87fae8a45bb87684 : SUCCESS in 37m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91e2c075a9ed4a3b8e35a22659c69ddf : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e88c1bbffc7f4ba1826c1f3c91fb7f53 : FAILURE in 24m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/767a6cf8410942c7859affcab42ecac2 : SUCCESS in 1h 44m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00e87481fc1c4653ba9a2787390acc27 : FAILURE in 2h 17m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b568ce497264e93af0f3706c1c051eb : SUCCESS in 2h 09m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c7f724fc6fc14f228c2905e060ce1f55 : SUCCESS in 1h 59m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d27354674bd3414b97dcbad0ed81c0f5 : SUCCESS in 1h 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6ff54e58ef9244e8826e17e0ad6e5238 : FAILURE in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b202680d6610380be2235e27909add2774a8d80","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 19:48:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a77b8e179bcf4c43558520b10545dd88af2596c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 22:01:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7516ad39d7e543bbba1814c11d2de27e : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b611eb1a27c94ea7a22e07e6b62478c3 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593161efbe9e4bb096a23714c4c55d72 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d201c3150a0d4694a71b85daba5cbca7 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c5627988d545c88ed4c678902cca14 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34a18a0ba8ee4f6ab36dd78ef4b91e85 : SUCCESS in 11m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/711cf791904f4d5f94ef415e66e751cf : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bff7cd818d244b1a9ccc64fe76304622 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76be5f452eaa4befacf9728432e57238 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f58bde334a745b5a949a0c100584eaa : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/adf19f8fda87427b953db9ac10126ccf : SUCCESS in 1h 23m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eedbc1ed5fe248aebcbb5b854a2be636 : SUCCESS in 1h 47m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f299d6cf3de24cd8a210c06740957300 : SUCCESS in 39m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/23cc0d7eefa74a049c7f244a85dbf4e6 : SUCCESS in 36m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bce21ada585424fa6d967c8f8a715a2 : SUCCESS in 38m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2652e2d217f4405bab46960efdf2e97a : FAILURE in 25m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce05261e857a4c658a6b495e0b8e7185 : SUCCESS in 1h 25m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b790c13b6885456097920645353cc775 : FAILURE in 1h 57m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/45759f50d0e04c5a9ac2f01162c3113b : SUCCESS in 1h 46m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6520f5704d744faaaef67b913187e816 : SUCCESS in 2h 06m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6935f402ac6c4eaaa7609c47e1d38142 : SUCCESS in 2h 05m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9971fd4e634941d48bee569ce18e2c14 : FAILURE in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d6358cdd225883e66dee2ff698bb9c7c9c8e499","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 22:04:40.000000000","message":"Patch Set 2:\n\nrecheck grenade\n\nHopefully this isn\u0027t a larger failure with grenade...","accounts_in_message":[],"_revision_number":2},{"id":"5e660214796b0c5e863e760800151f4a4c316a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 00:43:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ef965daf5b64c24b61c90d4a115adf2 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52034fefb7b34daa8da75071036876b5 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7204171eb034c8b92a83c678597f280 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33f0bf262ba148d9a91d9a7a33dcd248 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25ac2dc105fa4a9e808196ea08c0132b : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933afd48712e4e3991fe9cb9670ab63a : SUCCESS in 11m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e0f1fe3ade4449a85976afdb7565d6b : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/06f78bba3aac4476a9618af02885f1b1 : SUCCESS in 8m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b679b1fa33b4bb081539202db1b08c4 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82afd6c8caf54e269ad9977ca442e1e9 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4add4a72f1b74ec8b28747d466b462a7 : SUCCESS in 1h 23m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72f48ca81b574a3ca2edbee909bed975 : SUCCESS in 1h 58m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9a8e19eecc343cc91261802a377f8b1 : SUCCESS in 51m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a7e172f927fe4520b472fada15b5425d : SUCCESS in 45m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/901a1a4844394fbf8334a6cecfcad01f : SUCCESS in 42m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb1483fdc2324b78b913567cca1ca6ba : FAILURE in 30m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8e155ccf39b4a8e84f7725b5fe7dc8d : SUCCESS in 1h 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60324011faa5415a9e84ab1cba1561eb : FAILURE in 2h 00m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4f9c91d754f74253b5d87134562a389c : SUCCESS in 1h 38m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eeb61d59ffa54c05a4c37521469f06a5 : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cd1a0f916d92487ba88d9fec21c7c39c : SUCCESS in 2h 14m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/18034d26d7d5404ea464148083516c1c : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e995b441fdad6ba4f2ea3d0f8bada885f3111417","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-14 08:42:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"99ee8c5d28eb93f0bd60f2269673d38577bb7872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:05:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be198fb42680460da734d9c8fbf295ee : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/303081e8aaf54c1398589efc84895ae4 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4177d6f0aec247e39e1ee57e4ff85013 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31c9fd3342d24ddcaf96dc1e7bcb6541 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea3e2042d8c64a14a1aa8499c86dd789 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0ad615e5d14dbea5cb706cb07d6fd1 : SUCCESS in 19m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/46ba3b1fb3eb49789ecd21d6eece03d4 : SUCCESS in 8m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8eac809a17b3433c88f9d9e7fcdd254e : SUCCESS in 10m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aaa0ccb7c1754e539629bc59d97b7d85 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1158fee7a70b4b1aae0970ca5aebd6e6 : SUCCESS in 9m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c67809db61be448ea46b81d3fc9e00ac : SUCCESS in 1h 21m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88154487b5ae48e9b78577fe00f2ed2d : SUCCESS in 2h 11m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38bfa22e007446dcb78b29bcc7ff2565 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c760f6e32c24a1c8fcc118d30d5e89d : SUCCESS in 37m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aea25dc8d9994b6c9a206f340826b478 : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/75142d47009e4f7ba6cd13387a488b8f : SUCCESS in 1h 01m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/593501bb93e3472abcfa7826326b19f4 : SUCCESS in 1h 58m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ee42c1ff7934235b25a66b8d65c6fe9 : FAILURE in 2h 21m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e7a001ebbcb04f2583a71fb5882b0f40 : SUCCESS in 1h 55m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a03b68b60d024fc98d68b2be8172da31 : SUCCESS in 2h 04m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dc136e9a2efa42979b5dc93688d967e8 : FAILURE in 2h 00m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b72b91cbd47d485790a100ab7205a1c5 : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a42877ef7813a152abb0fb354e42a1aa733e773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:13:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f249291d052736c97e8ed47b14275f603dbbf19f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:47:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbae278772e24bdf9c82da2b1fb1f6b8 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c35c45aeea4f7c9a763752a4131ee6 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b44fa47877ed4374acc22a3f9e217a63 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e53a831e3af44bd68909535aeb776e4c : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c588f2b512dd434bbae09d4ea66d70ef : SUCCESS in 28m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3da9cd46986049958bb3f64a7bb65931 : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e111325c263d446ba0df042ee0b21b4a : SUCCESS in 13m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67760b4ef0b64d5e93d96f9c302af62e : SUCCESS in 1h 14m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ddb4c552a3a44eea075c14ace28da81 : SUCCESS in 1h 30m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac5fa847187143d0ac80fbd918a51fae : SUCCESS in 39m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f68afd15fd3943558bdd3df3fbc059a9 : SUCCESS in 44m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fb6433cea9b48f18c072222882591a9 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74586c0b218c4e31953c3731ff3b0bd8 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7d9d23aa624392dc34fb6c57ceb8a73015e159d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:47:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"983b83546413674aac0d100267f23d718504985d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 14:49:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2bf20db71541219802f8061b884919 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0b3086680f3db47e1494f96e6bae45f0ab24720","revisions":{"d893700c89ebff62df2e214dc9a17bb3fd83c0cf":{"kind":"REWORK","_number":1,"created":"2020-11-18 19:28:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/763249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/763249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/763249/1"}}},"commit":{"parents":[{"commit":"fc058c3a8367c369ab623083c34cb423f808c52e","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc058c3a8367c369ab623083c34cb423f808c52e"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 15:59:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-11-18 19:28:14.000000000","tz":0},"subject":"Handle listener not found during loadbalancer status update","message":"Handle listener not found during loadbalancer status update\n\nIf a call to update loadbalancer status cannot find the\nspecified listener, the API server will throw an exception:\n\n\"Error while updating the load balancer status: \u0027NoneType\u0027\n object has no attribute \u0027update\u0027\"\n\nChange ListenerRepository.update() to check the DB query\nreturned a record, and if not, raise a NotFound exception\nback to the caller.\n\nChange-Id: I6693be3e2d2f0e34f19f07f55b0429f8fda317b2\nTask: 41108\nStory: 2008254\n(cherry picked from commit 17ae2625da0487b7398b927e4e67a713b61569d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d893700c89ebff62df2e214dc9a17bb3fd83c0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d893700c89ebff62df2e214dc9a17bb3fd83c0cf"}]},"branch":"refs/heads/stable/victoria"},"b0b3086680f3db47e1494f96e6bae45f0ab24720":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-11 06:19:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/49/763249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/763249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/763249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/763249/2"}}},"commit":{"parents":[{"commit":"e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 15:59:17.000000000","tz":-240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2021-01-11 06:19:30.000000000","tz":0},"subject":"Handle listener not found during loadbalancer status update","message":"Handle listener not found during loadbalancer status update\n\nIf a call to update loadbalancer status cannot find the\nspecified listener, the API server will throw an exception:\n\n\"Error while updating the load balancer status: \u0027NoneType\u0027\n object has no attribute \u0027update\u0027\"\n\nChange ListenerRepository.update() to check the DB query\nreturned a record, and if not, raise a NotFound exception\nback to the caller.\n\nChange-Id: I6693be3e2d2f0e34f19f07f55b0429f8fda317b2\nTask: 41108\nStory: 2008254\n(cherry picked from commit 17ae2625da0487b7398b927e4e67a713b61569d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0b3086680f3db47e1494f96e6bae45f0ab24720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0b3086680f3db47e1494f96e6bae45f0ab24720"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
