)]}'
{"id":"openstack%2Foctavia~763437","triplet_id":"openstack%2Foctavia~master~I5d999d72cfe55600238c0daaa22766b7fe716e41","project":"openstack/octavia","branch":"master","topic":"constants-reimport","hashtags":[],"change_id":"I5d999d72cfe55600238c0daaa22766b7fe716e41","subject":"Remove re-import of octavia-lib constants","status":"MERGED","created":"2020-11-19 21:07:22.000000000","updated":"2020-12-03 20:56:47.000000000","submitted":"2020-12-03 20:54:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763437-constants-reimport","meta_rev_id":"266c7deb577e5c21ced142f9f3b79b073361a0e2","_number":763437,"virtual_id_number":763437,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 20:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-03 07:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-12-03 15:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-12-03 16:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 21:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 16:56:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 07:34:16.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"a2a193c9f1588512c34919dc54b3dddd13774f6c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-19 21:07:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91f97dc81d2045cf5e85a6c3a5bbafc99b1e8198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 21:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e09a1a3c97f409c9f551e597d990f99 : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3da846877854d72986a80684e3da201 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171927bca3434e53a461ab2fb295ecb3 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd106eb294d46d5b336abf78d6827cc : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/882cf3baeb9948f0a3e48fd0e6b0acea : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4689f05badad4c06b6204872727d9890 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb52907dff54774a613d269e5c436d9 : SUCCESS in 16m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2c3cb62f5f7f4e87a6a34186c6f2ede9 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/39b2301e0c4346a29cef9a923af6ff0d : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a317b07e31e435696daff6ad3af1cfb : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"597db7ed1560f742172b63dde581882a41d6c91d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 07:34:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74855e4d0731e5be168f19a1428cbbb3b7515708","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 15:59:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1b28c7551549da88efab62ace5b38287a9d5261","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 16:00:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"635a9f015b85d7a61a20eba0107b1ef582a6673c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:01:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee2d30f1b8f94acff5ec1e00069c23b5e4848617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:54:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88b8143fc99f452f82f2f9356e03c12d : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3b85d8ecff435a980fb9743ce32225 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6738ea1fe6c0442a9733d79a40826456 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/910d7c074ca049baa988278fdcdfa9cd : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b7fe69d9f841cc8723a9083d354602 : SUCCESS in 10m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a7a83b2dd0e41e8aaf9e300d960e8f2 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e1197d6fef049668cbdc98b520437e4 : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8f66b4fdf3747d9906a726b13dccbd4 : SUCCESS in 1h 39m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9526cdc18e5345438112da50f6d01b30 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/839cf7e63ccc4f8c94fdb095effda7fa : SUCCESS in 44m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2af98e63de2846238510c20b37191c0a : SUCCESS in 44m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bde530e00da047f9a20b902c270f2be7 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"53666d1cabdb6f7e9decfa06c8df587aa12e99ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:54:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"266c7deb577e5c21ced142f9f3b79b073361a0e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:56:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e52b19695842049f8754b159b2502f : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15a68fd426c3ca5de3302393f8352a20aab97fef","revisions":{"15a68fd426c3ca5de3302393f8352a20aab97fef":{"kind":"REWORK","_number":1,"created":"2020-11-19 21:07:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/37/763437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/763437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/763437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/763437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/763437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/763437/1"}}},"commit":{"parents":[{"commit":"58bf439bda1cdeaa3dac8b453e05485092bd2174","subject":"Merge \"Fix setting None in tls_versions and tls_ciphers in pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf439bda1cdeaa3dac8b453e05485092bd2174"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 21:05:05.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 21:05:05.000000000","tz":-300},"subject":"Remove re-import of octavia-lib constants","message":"Remove re-import of octavia-lib constants\n\nW0404: Reimport \u0027constants\u0027 (imported line 17) (reimported)\n\nconstants was already imported, just use the first one.\n\nChange-Id: I5d999d72cfe55600238c0daaa22766b7fe716e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15a68fd426c3ca5de3302393f8352a20aab97fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/15a68fd426c3ca5de3302393f8352a20aab97fef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
