)]}'
{"id":"openstack%2Foctavia~763477","triplet_id":"openstack%2Foctavia~master~I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e","subject":"Bump oslo.log version to 4.3.0","status":"MERGED","created":"2020-11-19 22:23:56.000000000","updated":"2021-02-09 00:53:22.000000000","submitted":"2021-02-09 00:51:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763477-secure-rbac","meta_rev_id":"87e3274fb1960bd9bdcfcf794d55fb82f3d3a972","_number":763477,"virtual_id_number":763477,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 00:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-09 00:53:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-08 23:17:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-08 23:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 22:23:56.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-20 00:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 14:54:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-27 12:22:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-28 10:31:14.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-08 23:17:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a5e9216b0b8b9ea5162fc7a98950cade421002e0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2982b8080cb3cc35a093a48ac2deee784b4b1b9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:25:43.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"e1dddb803be98d95fd4eb4f19adf288ddec2eb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/370e97ff33174cd4907cd2c7857612f3 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77c522c993b4b3fa01471aa263d877a : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d234a5aaf64c4b9fcb973506bc53d7 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73eee64fd08e487e865ca7773dbd7c27 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21e7208222ed4f49b8309536de44fa48 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab65f876270425689161261be313d8c : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c383473ceaf4a339ceb1ff5d82f754d : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90acd05e9d64eb29bc2a184f2fda862 : SUCCESS in 17m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d9bbc89f888400d8f585f05acedd23d : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dab3fa59be1042b4b9b7842eb6a8cf91 : SUCCESS in 10m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/912871cd6d344d7993af34c23d340e81 : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a292d2aa7ef4f50ad8bb877ab38d9c2 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f40443644d14689a7dcaaca77bf77a8 : SUCCESS in 1h 21m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d7952960ac14bf692215de3a82ad5b2 : SUCCESS in 2h 13m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df71eb9f6c6c4cc1af4e2fbc6bc0a6e5 : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/72e1518e79584cd4b1129a646c60fcc5 : SUCCESS in 1h 00m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5be1d8b38bb4ff48df19fc2a6287638 : SUCCESS in 40m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7883caf12d0b4442b5704e5b15c8d968 : SUCCESS in 58m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b55ea1d000434bc688296f9e5cfefcbd : SUCCESS in 1h 45m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5423dafc04604bfeb9e2c2735c65a0f9 : FAILURE in 1h 56m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/df2a986241774ac1ae2c6becaeea54aa : SUCCESS in 1h 47m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1cec6cdd61444a46844f18b057994ffb : SUCCESS in 2h 13m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2a8e3ad75fa2438d8670f5a7ef8d865f : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0a8beb2ac1d749b5878bd46a8d95a7a1 : FAILURE in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f096af03180b3f72eae93ec88a13b2d11ad3b33f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-20 14:54:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd52f410978b77a501dfcdaf7758e31a0c2f6349","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-27 12:22:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cc0a85246a3338db5b44e174c924ec13a4f3ebb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:31:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcfa5a305678d421ae6ed542a9028e6ac370ec2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:31:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e89f8d66edc833bb64f2689fcd304e0d3efce11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:36:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/be2cde61dc7d4b3d83372cf79f000085 : CANCELED\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6594026ce9f24d31a39c7f991a234416 : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c3f71f2a0441cfa5439b1696d2181c : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f55d2cbf51fa4d78ab61ff6bf5722ae0 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c7e26d4ccc14aaeaf50cf2f6e638a60 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b5414df3858b49fc94395bb03b12d6b7 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65cd053a56984e50be9c12e450cd7e2b : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c6abe0b0c9048cb9383a81f4b04bb5f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0f23c169114400dbf25eb7052c4eba5 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4ab5c0903cdf4027b4c084ca3b46c3fa : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bb9d88f31374870975b32b625ba7764 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9657fc8e3ee64d0a8df2ee9daa83bc97 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"580d05fd2629b73ab4db2413ad11bb420ca1dacc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:29:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7f6b52bec022559d4f9a4daaaa5793f1706e8851","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:30:14.000000000","message":"Patch Set 2:\n\nhopefully the rebase fixes the l-c issue","accounts_in_message":[],"_revision_number":2},{"id":"73833ab240f57f595b2ed7807a1b027f485237bf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 16:16:31.000000000","message":"Patch Set 2: -Code-Review\n\nLooks like this will need a slight tweak to the l-c file, and for https://review.opendev.org/c/openstack/octavia/+/752095 to merge as well which will fix the functional-tips job.","accounts_in_message":[],"_revision_number":2},{"id":"e69bdae5f8c3612ca24ac21f4f2e8b5620a90c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 16:44:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8fd31b6cd7c02e508b7a85ba062f759a1f57a76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 20:12:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7d240422502e3f900a905da02206cf8dff8f8afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 22:53:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8736eeae75a4ffa8df423b45828698f : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e3f6af2b2054263b42fb9a77338365f : SUCCESS in 8m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/599cca79f217449da2af283fab589e6b : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d890b1e6824f4c2abe21d88a8ba28466 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcb110b9aaed4488b69089d925ca8bea : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40c002efc0c46b88087c054be36b883 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07f80eb66d3f42c6be9c765056ba27ac : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405b082fdf6848bab5e850c034d186bf : SUCCESS in 10m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d8796e3c8b1847e2a7d98675eff1adca : SUCCESS in 17m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/71ad7dac23864dc89b612b0cddc9e41e : SUCCESS in 18m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/90dd85f02b454e378c26fa7babb2c5b1 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f46d90fb67c14027a9a3c674465dcf32 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5f659c89baa415b8416967d75b6cd42 : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b2f1e77f3ec4228bdffb111474e2584 : SUCCESS in 1h 54m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a4e91f3617f044cf8135abe4547a5e8e : SUCCESS in 40m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6632ea58d3a943a9a47de7ad96982e2b : SUCCESS in 36m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fc3acb06b5f4a98856c8f6b14277c3e : SUCCESS in 56m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/90af1a12b9c7472397817ee4ad2d1ab0 : SUCCESS in 1h 10m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4eee0f40b964ec9a11b8c4f21ad6219 : SUCCESS in 2h 30m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8d5ebb7a7b24f61bdcf5f412fec76db : FAILURE in 1h 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/17486ed27b034cbbbe28cda90198b4b5 : SUCCESS in 1h 29m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/66b4c1ced86e4ebca0f7ca0705587b0f : FAILURE in 20m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7c04a19b1198416da5eaca6398b6af0d : FAILURE in 2h 24m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/db8a9682db23496288a48a15bbe10a90 : FAILURE in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c21bdd91ef371f06b15daa913695f58261b80bf1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-02 11:58:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"852a34d3693a45c0b9a8b917a4904a6b21756bd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 17:17:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"29c158a765af116c8c4bc876bc3a3d8fe75abbf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:48:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/309bbdd7b64c4bdfa77bcae7213f5a9f : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/accc577f66d843d4905336a74be0fc64 : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f32f88e809a470180862c49d96fa36a : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7adbfd97dcfa42fa8b9c0e4229a3341c : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d37e125d00744c5ca7184471b6b6c356 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183b4368b7544192a7dfd88b34910bd5 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eb1aa87a5a64a11a303e9dfdc81ebc8 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e226778586146adbb426bfcc6597d56 : SUCCESS in 12m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/475cdeed4bdb45bc86c4fbd690d52ca1 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce931e24eed046beac88e8f885a135cc : SUCCESS in 9m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c600c62d4b714411b9b123e7bba598fe : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48b2bc8c2cfe46a2959a3f1c7198c882 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9cd9c04a66c485987b404aa22668d2e : SUCCESS in 1h 10m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b783bca4dbc4044b9208ecd15b3a89c : SUCCESS in 1h 45m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbcd8955d62446b2832a7f4d99eb86ae : SUCCESS in 45m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2b4e7fe7ed914337a30c9bca6993b7c4 : SUCCESS in 56m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0291b4b7311a479f8b4e3f937be826f6 : SUCCESS in 39m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a60750bbc7e649f5aca68d33e135d9f1 : SUCCESS in 1h 01m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab5acae28140443387df3996d026f793 : SUCCESS in 2h 12m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49e6f520d6554235a14f26656e45c114 : FAILURE in 2h 19m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/35baa81429cc47bf8adca8e6d459c786 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efabe2476a154994a157c2824a34921d : SUCCESS in 2h 02m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7738ddfceead4aada155641d404c932c : SUCCESS in 2h 13m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1595c481be704878955bfa6916def3ac : FAILURE in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d784f944e328a5a7986248fd02bd563e69b200d2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 23:17:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"844848d1df6b2350dbc030c8f3b295756f061779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:17:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3d3c35f84dcf0832923ed2a4455993e31a76f9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:51:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf7fccf5ec1142faa01b54d415f1adf3 : SUCCESS in 2m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e9012ed05145b8be5eba367c7e2ae2 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35974f9efb3498989d0f8085159c0db : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4a9e70e5a54efa8b9f7d220bcbf761 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8f2b581712f4c7b871b64f813814a2b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44e122c64f34ae6965c60ad7da2c56c : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0285001a5b8d49e7938aa0fb9ca2540c : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e888bacd7e4f40dfbd1fefc706f01fdb : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4dca003afa5549df9e71f23d55e20341 : SUCCESS in 1h 08m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b379ae24195d4ac49f8c86969b1a3710 : SUCCESS in 1h 32m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77c652541b39409b89b6efb52104d348 : SUCCESS in 59m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6aa206957f934386a1c2b7dcee8b5ed2 : SUCCESS in 36m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cddfb9ee17746de93b75006165543b7 : SUCCESS in 47m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6673eea6e95e4e8e8a13195802266732 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b5391b6b0f52a50cc9174ece6b106ead1187f169","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:51:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"82abeaf7f04d2b09d2802bba575b3084e0565357","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 00:53:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2205a0af6c2d4e9b88023eaf180efad8 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7","revisions":{"8302148018213b389e0d84384d802fb2af18ea89":{"kind":"REWORK","_number":1,"created":"2020-11-19 22:23:56.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/77/763477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/763477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/763477/1"}}},"commit":{"parents":[{"commit":"58bf439bda1cdeaa3dac8b453e05485092bd2174","subject":"Merge \"Fix setting None in tls_versions and tls_ciphers in pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf439bda1cdeaa3dac8b453e05485092bd2174"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"subject":"Bump oslo.log version to 4.3.0","message":"Bump oslo.log version to 4.3.0\n\nThis allows us to use the Wallaby release marker in versionutils for\ndeprecations.\n\nChange-Id: I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8302148018213b389e0d84384d802fb2af18ea89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8302148018213b389e0d84384d802fb2af18ea89"}]},"branch":"refs/heads/master"},"88fa9f1645fd40237cb086cfadb475f0795f2ffa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-28 15:29:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/763477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/763477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/763477/2"}}},"commit":{"parents":[{"commit":"fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33","subject":"Validate listener protocol in amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-28 15:29:47.000000000","tz":0},"subject":"Bump oslo.log version to 4.3.0","message":"Bump oslo.log version to 4.3.0\n\nThis allows us to use the Wallaby release marker in versionutils for\ndeprecations.\n\nChange-Id: I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88fa9f1645fd40237cb086cfadb475f0795f2ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88fa9f1645fd40237cb086cfadb475f0795f2ffa"}]},"branch":"refs/heads/master","description":"Rebase"},"8793b93900e747f5f46c4768d1c278fcb4f73610":{"kind":"REWORK","_number":3,"created":"2021-01-28 16:44:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/763477/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/763477/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/763477/3"}}},"commit":{"parents":[{"commit":"fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33","subject":"Validate listener protocol in amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-28 16:43:32.000000000","tz":-300},"subject":"Bump oslo.log version to 4.3.0","message":"Bump oslo.log version to 4.3.0\n\nThis allows us to use the Wallaby release marker in versionutils for\ndeprecations.\n\nRequired some other bumps to satisfy lower-constraints.\n\nChange-Id: I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8793b93900e747f5f46c4768d1c278fcb4f73610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8793b93900e747f5f46c4768d1c278fcb4f73610"}]},"branch":"refs/heads/master"},"7fbea7cb1ad336e6a84d27bcc588b961cdffd36f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-28 20:12:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/763477/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/763477/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/763477/4"}}},"commit":{"parents":[{"commit":"d2d5fc80f8274b7330c896b6825909ac04b653e0","subject":"Add ALPN support for TLS-enabled pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2d5fc80f8274b7330c896b6825909ac04b653e0"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-28 20:12:44.000000000","tz":0},"subject":"Bump oslo.log version to 4.3.0","message":"Bump oslo.log version to 4.3.0\n\nThis allows us to use the Wallaby release marker in versionutils for\ndeprecations.\n\nRequired some other bumps to satisfy lower-constraints.\n\nChange-Id: I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fbea7cb1ad336e6a84d27bcc588b961cdffd36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fbea7cb1ad336e6a84d27bcc588b961cdffd36f"}]},"branch":"refs/heads/master","description":"Rebase"},"7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-08 17:17:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/763477/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/763477/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/763477/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/763477/5"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:50:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-08 17:17:15.000000000","tz":0},"subject":"Bump oslo.log version to 4.3.0","message":"Bump oslo.log version to 4.3.0\n\nThis allows us to use the Wallaby release marker in versionutils for\ndeprecations.\n\nRequired some other bumps to satisfy lower-constraints.\n\nChange-Id: I3c2be8a5189dcda24c42ed7ab4d8fa33a03c5d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
