)]}'
{"id":"openstack%2Foctavia~763478","triplet_id":"openstack%2Foctavia~master~I29fee922eb112de08dd4f9f450942d3694b80cad","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I29fee922eb112de08dd4f9f450942d3694b80cad","subject":"Add common system and project policy check strings to constants","status":"ABANDONED","created":"2020-11-19 22:23:56.000000000","updated":"2021-04-07 19:29:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee925b53f362ca53e522d6ff352a2317bfd2cd2d","_number":763478,"virtual_id_number":763478,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-29 17:07:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-2,"date":"2021-03-06 06:55:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"date":"2021-03-02 18:51:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2021-02-08 23:50:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"date":"2021-03-02 18:51:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 22:23:56.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-20 00:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 14:59:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-28 10:32:19.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-08 23:50:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1f5f6042a9f2e56adf7eb268258a2f58dbcdb5b1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46bb2381ad34599918edbc14ba8f1a7b174cac27","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:25:52.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"ef80eab140dc46dc448dc7a5d166f6787ff431aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/262a667d2852415689a6c14a34f412d2 : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/223f757704934dacb03c9bcc55592aa8 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6da704c29942c28bd2ccf402787496 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f1ad6ea6584a9fbfdc57e7f444a474 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1420ca0fcd3f4dafa6c98241cd2d42c0 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0642e48efdc14d378be8953ca6d05f71 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ed970f99cd406eac6396d01c80aad3 : SUCCESS in 19m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b243a4f1126841748b6f9c1f9d621b20 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/56362d787600447bbea13dad9a57a3c5 : SUCCESS in 12m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ec7431e6036141ff94e9637e914d862d : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05c857cabdef4bf199e79ca2c3c5d938 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/262996f33bf64448a19dd6ba158c7ea1 : SUCCESS in 1h 24m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1e2598de92c4eddb176a549ae99dcec : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c927f3d67484f8daa1ed7dc29c0b53b : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7261699d22c44da384e47a1625fe2464 : SUCCESS in 45m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/542a07b2ff07448ebafc39c5678f2b1c : SUCCESS in 39m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/932753af74ff4fdf948ed0edaaeaaa82 : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85d5d702e1e54b2db21e1b569cb70c5f : SUCCESS in 1h 37m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/afe7cde0c96946889afc64059e2e6df2 : FAILURE in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4aed841bf3d745f2abaebf767203c7ea : FAILURE in 1h 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e17815485d524e8a92438b3a10212c31 : SUCCESS in 1h 47m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d04bacd818d64971bf4ec688ab760882 : SUCCESS in 2h 00m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/da197b18a7e04b1a8296fa20f8af2332 : FAILURE in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6de4840703c8e4eab2ad1e76290cf0284eadf8b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b13ed53cd4d4baae14ae6936fa3bb904cc59718c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:21:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95ace3e951784607a92708d3261c94e5 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcdb6e6d6c3d45fd80c806d59c012b48 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b732860dfb94b3ea647c1fbcaa493b3 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c35be58adf743c492e2bb8d3cbb546a : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/890292aa2a534496b901167738b43a87 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82674e165b74184bc30f2edbd1e7090 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bc456d797f40898c7d96d0e5e2025b : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/897738e3ed1747e0ab72eb008189de66 : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4ee9c14c71214e48ad622723f8c81036 : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9742b2dbb12b4c7c996114e3e05287e6 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d62bc4b7edef4d2bb93406b6e4fd7214 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7709f86772834711a480ab3020c66a6f : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8ab99be904a495db6ecaa670fc5368a : SUCCESS in 1h 59m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/441505dace74427d8c0a52106c994e6b : SUCCESS in 59m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/13d1af098aeb4ccbb1f8edecac1b2192 : SUCCESS in 1h 04m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28de5bb8bb324f14b6f40ec67833a3d9 : SUCCESS in 37m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c0125bbef024c4b9be0e6533654e208 : SUCCESS in 57m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/daf5cde40f364b398c7a9e310a2ba24e : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbbd5ab74536473585112f8a562cf1cc : FAILURE in 2h 14m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/16020fd0f3d44b149e8304bd0c3bfe42 : SUCCESS in 1h 42m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e10f41218b22468c8e1631a63be29714 : TIMED_OUT in 2h 34m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/508fde52518c4bf098c4acf407fe25c1 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f75eb9c7edee4f71a1f69ab2e03e9d71 : FAILURE in 38m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d96e2f97483c998297a12b2e42975dcab7cb598b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:32:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"af0d3dd851c5ab5a210acf2563758a382d5d4e28","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:17:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bfc07500d134a35865665294732b51e87c1ebc98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 23:09:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"98bef275e2a384206cc34ac165634f5a466e1753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 02:12:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d04dc4d8e440c297bed534e8b29b32 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25d56c0fba2a48f7a6ea9d0ead561e58 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44c42f9d4044413b1f6c854e603e3e8 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61d4563f04a2451cbd6cebdc1881f24d : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e78f7aa4caf4840a4bb7890212a5697 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d5c38d1eaa84914a6a26e9cdb7767e3 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d92834d603407ab9e3209abc77027e : SUCCESS in 12m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4a27eeff632e4eeba5f100ac68c50603 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0ba13f7babe04d14bf3a6f5efe7bba03 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f0ad661509bb439787819f5647ee6be8 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65848e1965dc42fda7add4fd09a8a93e : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4abe5ba006474db2934d8dcf8f3ae8fe : TIMED_OUT in 2h 13m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c143c0df876a4ce984a35ae74ff3a13d : SUCCESS in 2h 06m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/371669e270f84e518f9d5436ff115bbe : SUCCESS in 45m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9ba90baca4a04282a89ff60f2e5a7374 : SUCCESS in 46m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0850aad50cab4d1d97b366d60c3b39fc : SUCCESS in 44m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ace6c3b54b64a5a9b8d75c7be6671ab : SUCCESS in 56m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c845b432f9844b1a6ec799c5c27ddb0 : SUCCESS in 1h 49m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/509857ed3ca44266bab7b23b18e38333 : FAILURE in 2h 06m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7be41c7e343c4d659f5532604f5a97ee : TIMED_OUT in 2h 38m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2bd0abfe86e84067b1864a0d8ae62062 : FAILURE in 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1565c48974bf4f9bb53efce882143731 : SUCCESS in 2h 03m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cf16646f73be4e4db785ecd615f53127 : FAILURE in 32m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"363a151466e7c9d5059a3770e1759d3eaa1a2cc2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-29 14:57:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e4e4f3f06bd8824e82b70847629beb2aa1f77592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 17:07:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c1aab1f6fb1402e9b79f6b8f68f7e32 : SUCCESS in 14m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a421bfd7294424582038caeed43e8ca : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff78ebade1fa4f808f6bfce088fcff10 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bba86944a4249d794820bce74865a3c : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed3105ae61b40de8c8252ab81fcb008 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8c2028302864b08a97c9ac888ae09a1 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2019c86accb4455b9b7b919955f3ce17 : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de1bc97d6ff4491a9c4c4f46805892c5 : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e69fadee7ab5458e95c6117f94b0fe8a : SUCCESS in 8m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b188c19703944cf9cf220d6fa254233 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a16b7318e177460ba7940b2645c25e2b : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aff62e80f63c4d44b9961fe1675f46d8 : SUCCESS in 1h 14m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6463b459400e43ad88cf6a4461eae23d : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d0a66b4f70c4ebd91736aa504683270 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dec5259e461f47f4b0f713da5185b2b0 : SUCCESS in 40m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/708e5c6299fd4d1c96dcd9cd20b090dd : SUCCESS in 40m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/653306ad3d0b4079ab8d26f1b92a4698 : SUCCESS in 1h 43m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d80abee51f444de941ba58c68217292 : SUCCESS in 2h 02m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c7218515b6742d68e02ff66112c12f7 : FAILURE in 1h 56m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f3b32039cb82477cbc753ca08c538d83 : SUCCESS in 2h 04m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/18582faf02384404a5c2a7fa7878d50e : FAILURE in 20m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e9315f3cb7cf4e07af9247958f93424d : SUCCESS in 1h 59m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a1bd48072e95417ea4568375b9cf6c27 : FAILURE in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d3445404e8b615e08d626a88b06c76591c91736d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 23:50:46.000000000","message":"Patch Set 3: Code-Review-1\n\n-1 to hold, these are incompatible with our existing RBAC strategy. They will need to be corrected.","accounts_in_message":[],"_revision_number":3},{"id":"f8cc5cfac0564c1a0536a1f2746645deba6a726e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-10 17:54:28.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e -1 to hold, these are incompatible with our existing RBAC strategy. They will need to be corrected.\n\nHey Michael,\n\nHow would you like to see these updated?","accounts_in_message":[],"_revision_number":3},{"id":"86490e9daba408ee2123d9d0a26c78a4173ca92f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-10 22:14:31.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \n\u003e \u003e -1 to hold, these are incompatible with our existing RBAC strategy. They will need to be corrected.\n\u003e \n\u003e Hey Michael,\n\u003e \n\u003e How would you like to see these updated?\n\nHi Lance, I am working on the Octavia RBAC this week. I will make the required updates.","accounts_in_message":[],"_revision_number":3},{"id":"27f11451ea8bde58edb70ade448c7b9b2f0e1448","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-10 23:06:04.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e -1 to hold, these are incompatible with our existing RBAC strategy. They will need to be corrected.\n\u003e \u003e \n\u003e \u003e Hey Michael,\n\u003e \u003e \n\u003e \u003e How would you like to see these updated?\n\u003e \n\u003e Hi Lance, I am working on the Octavia RBAC this week. I will make the required updates.\n\nSounds good - let me know if you need any help.","accounts_in_message":[],"_revision_number":3},{"id":"699f316e2cdb970025584a11e73240494a0dfb98","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 02:23:22.000000000","message":"Patch Set 3:\n\nPatch https://review.opendev.org/c/openstack/octavia/+/775957 will likely replace this one to maintain the Octavia Advanced RBAC capability and backwards compatibility.","accounts_in_message":[],"_revision_number":3},{"id":"7b1667bc8157f3d5367a57bb14ee2685e17828ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:56:59.000000000","message":"Patch Set 3:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":3},{"id":"4610f8e04772ea3d916248c8776ee72effbefd9e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 18:51:22.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"51b1b91d3538f35f8cc52c0642e577d70049661d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-06 06:55:38.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"b271f32c8abfb9d51967d34241456632c418ea6d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-06 06:55:54.000000000","message":"Patch Set 3:\n\nLet\u0027s go with Michael\u0027s approach.","accounts_in_message":[],"_revision_number":3},{"id":"b60731888f415ef267981c47677fbadbf05736f3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-06 15:26:29.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Let\u0027s go with Michael\u0027s approach.\n\nSounds good - are there tests some where I can start looking at?\n\nThanks for the reviews and discussion.","accounts_in_message":[],"_revision_number":3},{"id":"ee925b53f362ca53e522d6ff352a2317bfd2cd2d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-07 19:29:09.000000000","message":"Abandoned\n\nAbandoning as the team merged https://review.opendev.org/c/openstack/octavia/+/775957 in Wallaby as a replacement to this approach.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bdb6c63fc097a1590d8726596eed8bcd08d71160","revisions":{"89390dc2810fb53e6874ce0d800fc2df819c5e1b":{"kind":"REWORK","_number":1,"created":"2020-11-19 22:23:56.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/78/763478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/763478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/763478/1"}}},"commit":{"parents":[{"commit":"8302148018213b389e0d84384d802fb2af18ea89","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8302148018213b389e0d84384d802fb2af18ea89"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:54:49.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:54:49.000000000","tz":0},"subject":"Add common system and project policy check strings to constants","message":"Add common system and project policy check strings to constants\n\nThese are common policy check strings that we\u0027re going to use to\nimplement a few basic personas across OpenStack APIs. This is going to\nhelp move OpenStack towards a more secure and consistent authorization\nexperience.\n\nChange-Id: I29fee922eb112de08dd4f9f450942d3694b80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89390dc2810fb53e6874ce0d800fc2df819c5e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89390dc2810fb53e6874ce0d800fc2df819c5e1b"}]},"branch":"refs/heads/master"},"e74135501ff29c51e2d111e0a3809ddf65040389":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/78/763478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/763478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/763478/2"}}},"commit":{"parents":[{"commit":"e9ae51610e1d6b659942b926dcb49ad15f30f530","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ae51610e1d6b659942b926dcb49ad15f30f530"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:54:49.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Add common system and project policy check strings to constants","message":"Add common system and project policy check strings to constants\n\nThese are common policy check strings that we\u0027re going to use to\nimplement a few basic personas across OpenStack APIs. This is going to\nhelp move OpenStack towards a more secure and consistent authorization\nexperience.\n\nChange-Id: I29fee922eb112de08dd4f9f450942d3694b80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e74135501ff29c51e2d111e0a3809ddf65040389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e74135501ff29c51e2d111e0a3809ddf65040389"}]},"branch":"refs/heads/master"},"bdb6c63fc097a1590d8726596eed8bcd08d71160":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-28 23:09:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/78/763478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/763478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/763478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/763478/3"}}},"commit":{"parents":[{"commit":"c6dc0878a661d6199a92ed14ecfefc0adf58ab07","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6dc0878a661d6199a92ed14ecfefc0adf58ab07"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:54:49.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-28 23:09:55.000000000","tz":0},"subject":"Add common system and project policy check strings to constants","message":"Add common system and project policy check strings to constants\n\nThese are common policy check strings that we\u0027re going to use to\nimplement a few basic personas across OpenStack APIs. This is going to\nhelp move OpenStack towards a more secure and consistent authorization\nexperience.\n\nChange-Id: I29fee922eb112de08dd4f9f450942d3694b80cad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdb6c63fc097a1590d8726596eed8bcd08d71160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdb6c63fc097a1590d8726596eed8bcd08d71160"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
