)]}'
{"id":"openstack%2Foctavia~763479","triplet_id":"openstack%2Foctavia~master~Icdbe170847652952b994571c54df51f0c9a1b349","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Icdbe170847652952b994571c54df51f0c9a1b349","subject":"Implement secure RBAC for amphoras","status":"ABANDONED","created":"2020-11-19 22:23:56.000000000","updated":"2021-03-10 16:44:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70bde12bb21be3d655110f321be025bacbf02acc","_number":763479,"virtual_id_number":763479,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 06:47:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2021-03-02 18:50:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":-1,"date":"2021-03-02 18:44:30.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 22:23:56.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-20 01:01:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 14:59:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-24 21:58:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"57e34e15dfdc8cefd572166bdcf901a6e1c85efc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86659faa1cfed13caaa7dc6b00ad18f5ff4ac798","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 22:25:56.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"82c467b78b4db5ceb477b38d91539c7ab5355e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:01:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10d510fb020245ddb42dfe9157bffe83 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4173fa2187464f659a52058a777a036c : FAILURE in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c77f31b751ab45dd94e9002c733856aa : FAILURE in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81584470ed1d4246a005a7e8bf3a62ec : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b736bffe7c3f49729156592ad37a52de : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2efa57e0fff0431390351658a1077579 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493d7f9129e147aaa456cfd2ded46029 : SUCCESS in 18m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08c692dcb3a146379b5b7c025cbef3ad : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1760d8e1c6974a2f8158ceec2e1ca1ab : SUCCESS in 7m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a7d1f41ff864df6b1d24f931297ad98 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/146e349a1c5f4c0b89c8283cab844a2e : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a239cc96e4346c2b725323555fa9161 : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23838c511b84485a8ead4a327925e369 : SUCCESS in 1h 46m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60f1cac6a7c54d8f935e687971ee9638 : SUCCESS in 1h 03m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2b10ccdd6e7c4239abedc015f697e02b : SUCCESS in 39m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3347eeb4c314d41899f593e55108727 : SUCCESS in 40m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5494c535f4724fb290c8744881efa448 : SUCCESS in 59m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b658607da2634c6abe8c0e474c204d7c : SUCCESS in 1h 39m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7cc005d6b44b41ebad20d52ba25a5744 : FAILURE in 2h 17m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc657a61de064dbfa94ab9192473ca46 : SUCCESS in 1h 42m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f843a6301424130b5dd7a7b2e6f7505 : FAILURE in 2h 15m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d7a16ab8376c4363b6c4d83cd9612383 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d35481c0941d479bb7fab747fbfca1c1 : FAILURE in 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3f25f827d8f844ab348ab8490de9002ee1a7d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbf1d8910b5096f498a61803c1508fdd9e54cbba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 20:28:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9fde0ae755f41a2a8c54bd7bfea2431 : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4fdc514ae174a1ab2da307fa75d63c1 : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ede31ee53f495180afc0859205e96a : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d19a93c0045346e794a1f286e57ce781 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84ee8a7e0aac4b3b8d80094308b9fd3c : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c961f72c8b74abdabb5abbba9f2e37d : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3baa21b55994216b675287738dcfebe : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d9b789d3d0474194ab246a5545605a13 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0cbfbcc84cc48d6a30f54f1af077956 : SUCCESS in 8m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cddea60ce274d07abd8c8809172b19d : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aebfa4e5e9374d91ad16c0554315d26d : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffd0bf27de8e44fb9962bffc542831d3 : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0191415083c543d68e29a002459fb5d2 : SUCCESS in 2h 23m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d64deaef24a84bb69ec6f8d50044dfbd : SUCCESS in 41m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9a27bec409b6438dad37a8499cabc1ef : SUCCESS in 38m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edad71262f5040378e6f93064565a46e : SUCCESS in 49m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dea6642d5c834a24878584cffe73982d : SUCCESS in 59m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b26e9e25d4bc4113b4db0160a04debb1 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2a358eaf70445d6a3ca69968406e144 : FAILURE in 1h 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/992c3006c5b04fe08eec9e5ac4a99884 : SUCCESS in 1h 38m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2453b83c38664a35bae7b38db93b67a8 : SUCCESS in 1h 51m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/db923c6c4c3747c3b5879fcc8d49a89b : SUCCESS in 1h 56m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fbe2e2e7c1564945beb1e41b7cc06fbf : FAILURE in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13a1bdbf158aabf7e8a810ed4024b2fccd62503f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1c59060154d4c2833438d7c0a21cf1f3c0714d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 06:47:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d275756faa744059b659552ae2499739 : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f642e29e42e4523be013d83fd19c6b8 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2faf641e9b4041b491f6e67205b0350e : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b85e3ca5572473c9e467cd92c2006ee : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd8371b03d04653aeac6f343a508a3f : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3d53808837849529edf5ccde2545ddd : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acdb527adef41d68e7710203d1956b2 : SUCCESS in 13m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/20cbdc204c4749019546e1773f1294e6 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/71350fed248441d5a7c69a15ac7bd5c5 : SUCCESS in 8m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c2e15f9c28e44b869474e4a9c012ee1d : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/534188f8aaf64f20a491494cc43ea1cc : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cad88778a65744f7b04cd011fddd89b6 : SUCCESS in 1h 36m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb2a8d3c79d14e5bacd5148fc87eb488 : SUCCESS in 1h 51m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/69d7e1f187de429897b68dda92f70fa3 : SUCCESS in 1h 02m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1621945aaad6477397bfd626089b1103 : SUCCESS in 37m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e617f6623074726ae18b892da0b0421 : SUCCESS in 38m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3e2d6c37dc343cb91ec4c6382264ebe : SUCCESS in 1h 01m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95a2fbff19ce4c2ca3750dd1e66b56dc : SUCCESS in 1h 52m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b43adae36d14dd38fea3bdc212236af : FAILURE in 1h 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/532ca83631584e0ea5a357e1b6a5d3f5 : SUCCESS in 1h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d3b9bfe1bf64132b95a81407b3f34ee : SUCCESS in 1h 43m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3a8094a0070447d7a475bf1d40cd1378 : SUCCESS in 1h 55m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/124744f72e7741d28f14f3b12a99bb90 : FAILURE in 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e5036cf6e1c85d4c3152cc01b23bcd57b18fe75","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:22:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c5dbce0a7947a7c0019b405ad6c2b1c9156f64bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:58:09.000000000","message":"Patch Set 3:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":3},{"id":"5465aa071cec14b52d5623ab3c7aab42dfa5c8c9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-26 02:52:02.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e An alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\u003e \n\u003e Please consider if this is still necessary or a superior approach.\n\nThanks for the alternative approach, Michael. What you\u0027ve proposed is much less invasive.\n\nIf it delivers the same functional behavior as the personas proposed here, then it\u0027s fine by me :) \n\nJust commenting on this patch, but I see you updated the rest, too. I feel the same about those as I do here.","accounts_in_message":[],"_revision_number":3},{"id":"dea97ef939ce18a683c8a3e429505cae427ebb5b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:44:30.000000000","message":"Patch Set 3: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":3},{"id":"eb8734fbd716c9231c9320f6cec6e1b28a74576a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 18:50:55.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"70bde12bb21be3d655110f321be025bacbf02acc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:44:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"580b5ce43bc9055be2262e3a0fbf46b9733c1f16","revisions":{"cb3ab220ad23942083cbb77bc45c0d4c5b7e021d":{"kind":"REWORK","_number":1,"created":"2020-11-19 22:23:56.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/763479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/763479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/763479/1"}}},"commit":{"parents":[{"commit":"89390dc2810fb53e6874ce0d800fc2df819c5e1b","subject":"Add common system and project policy check strings to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89390dc2810fb53e6874ce0d800fc2df819c5e1b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 22:10:46.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 22:10:46.000000000","tz":0},"subject":"Implement secure RBAC for amphoras","message":"Implement secure RBAC for amphoras\n\nThis commit updates the policies for amphoras policies to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Icdbe170847652952b994571c54df51f0c9a1b349\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb3ab220ad23942083cbb77bc45c0d4c5b7e021d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb3ab220ad23942083cbb77bc45c0d4c5b7e021d"}]},"branch":"refs/heads/master"},"d80f5d98f8866fa36e3dcb2b819414a5f9075aea":{"kind":"REWORK","_number":2,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/763479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/763479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/763479/2"}}},"commit":{"parents":[{"commit":"89390dc2810fb53e6874ce0d800fc2df819c5e1b","subject":"Add common system and project policy check strings to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89390dc2810fb53e6874ce0d800fc2df819c5e1b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 22:10:46.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 14:34:35.000000000","tz":0},"subject":"Implement secure RBAC for amphoras","message":"Implement secure RBAC for amphoras\n\nThis commit updates the policies for amphoras policies to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Icdbe170847652952b994571c54df51f0c9a1b349\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d80f5d98f8866fa36e3dcb2b819414a5f9075aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d80f5d98f8866fa36e3dcb2b819414a5f9075aea"}]},"branch":"refs/heads/master"},"580b5ce43bc9055be2262e3a0fbf46b9733c1f16":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/763479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/763479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/763479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/763479/3"}}},"commit":{"parents":[{"commit":"e74135501ff29c51e2d111e0a3809ddf65040389","subject":"Add common system and project policy check strings to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e74135501ff29c51e2d111e0a3809ddf65040389"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 22:10:46.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for amphoras","message":"Implement secure RBAC for amphoras\n\nThis commit updates the policies for amphoras policies to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Icdbe170847652952b994571c54df51f0c9a1b349\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/580b5ce43bc9055be2262e3a0fbf46b9733c1f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/580b5ce43bc9055be2262e3a0fbf46b9733c1f16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
