)]}'
{"id":"openstack%2Foctavia~763481","triplet_id":"openstack%2Foctavia~master~If51065088bd7c5c0189cd1b13f49ef7f6089afb3","project":"openstack/octavia","branch":"master","topic":"master-to-primary","hashtags":[],"change_id":"If51065088bd7c5c0189cd1b13f49ef7f6089afb3","subject":"Use more inclusive language for amphora roles","status":"ABANDONED","created":"2020-11-19 23:01:04.000000000","updated":"2021-07-28 20:33:17.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"949d84d4219f343cbb523f51e9658907c1a7fb97","_number":763481,"virtual_id_number":763481,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-02 00:11:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2021-02-01 23:10:21.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 23:12:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-20 00:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 21:46:22.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2021-01-14 08:50:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-01 23:04:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"fd7adb6511f900d138fd58322beb3602d8251567","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-19 23:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"502d70f428eab54a5d8ed9a51e820aeaa89a426b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-19 23:01:36.000000000","message":"Patch Set 1: Workflow-1\n\nHad some questions in two places here","accounts_in_message":[],"_revision_number":1},{"id":"8f8a4bd5879b86d7d36179cbecd7f4f245895a22","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-19 23:06:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"332ef64ee2e87dbb0a98fc4b9f2dbd0cc2e3595b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-19 23:12:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6a37b1040de47fdf306f24665d55b03ff9003c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:02:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0abdb366708140c0891bcda1aea7358a : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f0686e5f7064a6c94a3340b3611da97 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e84f66277a34ff7bddec86f8e356582 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccba4dbe634f4c5b9f7f4cdeb69aac44 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5c1b6272a344836b1c9f11b08374266 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/672e8085f5b143a59a53d40ffed2359b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324f83ec2e274617af85e68ed5f02216 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b26e9173d94eb2b01375458a9039e0 : SUCCESS in 19m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b98a00e0c5d42078e973164be68f7e6 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a406eb0b2c54567af1826b0a41a548c : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c5eeeef7c904d70b9600e9b226dab38 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47bce0bd5a754b4bb7cd992de4f20cbb : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf0a6d3c3e474358a75fe422c2546fe0 : FAILURE in 29m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c4f43c5be014df698248987c4806c35 : FAILURE in 21m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33a1b2cf67ac4543abaee5bf85a47e80 : FAILURE in 22m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/025163e5c75d44b5918dc26c3893807b : FAILURE in 17m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e79e90ca5e284e66a351111fe1661abb : FAILURE in 21m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c79a724b225f409292efdaf0c3505732 : FAILURE in 49m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c376ba4168ba497887916b83ac808bf5 : FAILURE in 24m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60bef5b2a56747dc80f637273b2fa410 : FAILURE in 24m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/879a3a386a964251bdea8bc3f7775cd4 : FAILURE in 19m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ad203d3ecec42d5971b75883290db63 : FAILURE in 20m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dbf603908180446e8edf1b4eec1ab2a8 : FAILURE in 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3e0be995ba8b4978b2914f578d24cb29 : FAILURE in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84f4705beba923004b579162987306832aa5fddc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-05 19:54:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"704757359f404b8d277fac4e263ce6783f4ae5f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 21:05:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2cd6788090c64d14a73fdac650a674ac : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3166b7c9410540e3834f2d924d4af5c0 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03335fe980a143f3b9fc8487bb1b3a31 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09e79e0e6f042ccb28e319c05146986 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f574278d334b4dceb4499e8233184258 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a81a78d58e6425d9f971812a4cb6ea0 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e2bd23fa0ef4908bc1526797bf1b100 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af69d6d98ad148908ad83a69974482f5 : SUCCESS in 15m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2ab615f9e48e4b78ae31f1e772e7edb9 : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/836d377071ea4cf0b67701561dcb31a9 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b7271c08c024cdfa07e10ddde7c1290 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fde6fd876efe4d4cb196cea2110b6498 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/861666a8ccc146b3ba735063bd2a28cb : FAILURE in 28m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7d87967a7be4047b48936a1d0a7e109 : FAILURE in 18m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1649817b1f444e2380f48d853565503e : FAILURE in 23m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5346c8b5b9f5434fbc820c683b55b78a : FAILURE in 19m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98df288e59944b7ab3da879920254198 : FAILURE in 19m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/94d7f42108ec4404b6a03a8b0f3482a8 : FAILURE in 1h 06m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c930b46eb7da488796967d3e5be86a4d : FAILURE in 11m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/535774e8b4794fb59fa279097936917b : FAILURE in 11m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/046c903bc9674e2d9f021c9663d096c4 : FAILURE in 39m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcbadd0ccd4a4150971970f81f85ed2c : FAILURE in 41m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0a020ecf4ca14d32a3fdc15447483aca : FAILURE in 21m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7c5fe3b5856f4668a9a03c185e12f391 : FAILURE in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e62401bbd1b191eb29d566cba0148db45af64f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-05 21:30:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9f89ae8dac7103b4b9e0c07408626259b262322","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2021-01-05 21:46:22.000000000","message":"Patch Set 2: Code-Review+2\n\nThank you!!","accounts_in_message":[],"_revision_number":2},{"id":"6b27f60496cd8a19c0757fdcd11cc4e1c1e3bbce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 23:44:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/342b03c103cc4c60b3ad945b90be3dd9 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32fc982622ab4b6394b20c9769958248 : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f24f529a5874c478f1c41d00684d4a8 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70cd21be4336467ab29aca8a1a0e31e6 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c960e49b143d48a6a14e8c2e2aa7584f : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71943b89a1e148eaa69e917cc98ede1e : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77268e49724c4b8f8788c689cec5b1e9 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b7634fef0b48eebce94d5800fcb0f0 : SUCCESS in 13m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/807403d715ea4a419bf8b4d584166093 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/045036585d36456abac377cf5772dc36 : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8e99e386dd14446a3808eac2a73c0fe : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48ba0c744036421986e48c7d55633bd9 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7992a7d4d5e44aa6865c205f9d75c903 : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d079bca3d73e4557a1087acb4f4d420e : FAILURE in 9m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/46999c3702bc48d6b643a3853ebb525e : SUCCESS in 56m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/482d639e1856498da40ed5471c5f4ba7 : FAILURE in 8m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09de514320394cfeb41e146e682a8d8b : FAILURE in 9m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2de8d4de9c6749978befb409aea47189 : SUCCESS in 53m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cae30957299d4123874613904a110e8e : FAILURE in 9m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a005b89aed44c46a8cd728fbddb3edf : FAILURE in 12m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09505c8fa82d4bd79dbcb65ad51b289f : FAILURE in 8m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ec9a902cb8c4c8d90c816a3def9efe9 : SUCCESS in 1h 35m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2f8100e609ac40a98bec880a7a9a3a6e : SUCCESS in 1h 56m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6a88187ab71343f4acc602785566f9e4 : FAILURE in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5aed992b30f9501fe9392974c8238518ff530bd5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-06 03:18:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1f5dff4e302866227c72fe8b103433edcd9db905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 06:34:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49025e5217f9449081a3dc6a204f6015 : SUCCESS in 5m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e3339d709243e18eaafd35cea3f7e2 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d39114f31cd4566ae9d29875b870301 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b3dd4a12644154a5d543233f01e5e5 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a36f387d8ae4881a11039887db8cc40 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ffd03b61c5049faabe726aecceb7924 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67619b26381841239e02ea39c0e27e2b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b02c6fc7101441c840d00230feda10c : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/812196c231104c1a99340d84fe21c6cc : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/672e131d1ab848c1afac9793c9cb13a1 : SUCCESS in 7m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1378454b9e2b407192b3e1aa7c8b318a : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa74b7fea9b34785807bef15cb3202e6 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6cc1295e71eb4a5891a5aea2c954f628 : SUCCESS in 1h 15m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d91a91474e274aa5a99e30ed544bfe49 : TIMED_OUT in 2h 36m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0986c669f984082a41fdddad49087ac : SUCCESS in 38m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/24cb41bfc13c41d3af3387a9c38f3f76 : SUCCESS in 1h 02m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef9ead2504e042cda594b657f91f2dce : SUCCESS in 40m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59ccb03be91d4a6887f77016207d6267 : FAILURE in 10m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77bdfcb541d145f48c64fb7ef77ea903 : SUCCESS in 1h 31m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98d7a38836fa4af8b96c92f929e240b3 : FAILURE in 2h 22m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c29cbd976eb94c44ad6995921bc0f7b6 : SUCCESS in 1h 36m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aafd3078696f4896b1b4e02c49b96fe5 : SUCCESS in 1h 28m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8aa4bfb9aec546188f197835610bd392 : SUCCESS in 1h 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/20969a04b8b4404991d0301998551422 : FAILURE in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b8ddc7021df967ef3e79fc3f19a5ba63978b4ee","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-06 13:38:59.000000000","message":"Patch Set 2:\n\nrecheck\n\ndpkg issues","accounts_in_message":[],"_revision_number":2},{"id":"10ad4b174466d97f72542bdcba4a9c015e66282b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 15:59:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7da98d978ef74c1ca4bb0b532089807c : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfaebcfa50834096861060bc46c34e8c : SUCCESS in 8m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa06ec8a3b574d168e5c1cc6dd75091b : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1956f242474fdd8083dffa4b5ac805 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/163d3fdb5e1448aab82f0e9116a12566 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a5dd04d3b4c41448e20a44838b2c1f4 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39b00de51a1448c0afeca12d09322695 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e3ff8a023b4aed9ce3ada611394c42 : SUCCESS in 16m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/42abcaf63ccd4808b3b6c6f7e343b405 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c071af5d97cf4b828525d26769f4fba6 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b7db58e9c7a497b8c0166c4ab662906 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65d6ad43701b4e2bb64e0721df8c06e0 : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c1018ff7cbb4e43b35cd1f779e3fef9 : SUCCESS in 1h 17m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db4aedbb7b4649dc91c1ba22db77a481 : SUCCESS in 1h 36m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a91703aca4dc4ec98450564ac0092d90 : SUCCESS in 38m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f9cf72e6c9e46ecb9028c8239b7261e : SUCCESS in 40m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d33a0cbed52e4f55bec527982dcf31f5 : SUCCESS in 48m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38751fb5f51f4b1497b73a9db4fec888 : SUCCESS in 59m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a8ee51d729d4eaba71a0dd31fd65c0f : SUCCESS in 1h 50m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b623a6a6dee468e8ec9ebe0e340ac2e : FAILURE in 2h 18m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/108a72f7ece54f80bbc141b188dea6cd : SUCCESS in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0861cda38c8a47a69d87a7981e69e6b1 : SUCCESS in 1h 58m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/170f2c9548844ac6be440682c2bb7036 : FAILURE in 1h 43m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a8dc97ea6fa349a1921f0ed4ff00de6a : FAILURE in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"938ab6b53ea11bc2ef91f0517e30053f146d388b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-14 08:50:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nI think we need to add release note for this.","accounts_in_message":[],"_revision_number":2},{"id":"2471af4f7bb433ede1172232d7814ceafb4ec61a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-01 23:04:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d38fcb38d65a3b8edc9b920b56e0e87a4d1682e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-01 23:07:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5b723cef8145603dbb2a05487a122b68c19578a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-01 23:10:21.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"def3d740595e91001ca6621429f85e7d3e835534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 00:11:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ae64d14f84124f91866d02e5d0d3ce2c : SUCCESS in 5m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5522411f7676459ca61d2672f5e537fa : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d64d6a00a0d4023961d0f8b9e1aa478 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43ac96e73cd475697da454de4e3ef8d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2d8cbb3c18342f896e7e22b908b3be1 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332c1e88553c4779b948d1e021acc0ac : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf11d83480804457bb49743618568f4d : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350063712cf54b1eaffd3fc0436e9d7e : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f44796a1879a4ec9a28c4184b7efd864 : SUCCESS in 6m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f48b3fc1aa04c53973eb1bd8876eb53 : SUCCESS in 9m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b02745cccafb4bbf930ec9f9f2814daf : SUCCESS in 8m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/14a4e67edd684bf2801725af91844bed : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e6b41d169db42ac8055c4abdf528512 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/75c1d40e8fe24f22a55e4551d2515123 : FAILURE in 25m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/180f2f69ca1a44418c6252b5c166f3f5 : FAILURE in 24m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcf829162cae4a87bc6d74d90daf9dae : FAILURE in 22m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/13e8d1bf53d142a893ab19e85a739d81 : FAILURE in 22m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b6d12cc4a6e45be9e82df2bdfdb0fd6 : FAILURE in 18m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3593a4b3630144bf8e5b51b1fd11ac0a : FAILURE in 54m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8529fc86f51445ed91e9b8336be54f03 : FAILURE in 23m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88b21e90c7804fdc982e195205b3ff63 : FAILURE in 21m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e6bb90970ed644c5a7d75728b2e33e7c : FAILURE in 58m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e0dea24a26d7476595b40cce487d0872 : FAILURE in 51m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/51664d20ced24768ad6ddbbbb5a71308 : FAILURE in 51m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d51e80e554b74a379685093b97656fcf : FAILURE in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ba155e478428264b94439c7cd3d473aa5c2aab0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-02 01:35:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"68d16ed01c607fdfe2af2f3c64446585933b990b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 02:25:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/83fb6e1356d04741a1c1bf57868726dc : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ba23a883f047a29a7811e871c46d5b : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61f3253fcd6845e69da98d148d6e3def : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108c702189034631a5ad05af4facc2c4 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5adc9d962a94037a4b2e8d4ff48695c : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54df838a31bf41379e3adefa3788d9a8 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7732d77a84241f78df1f23899a87444 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e574b3a2314f59ba51cc12333b2681 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa5c88a376274cfc925d8b56cce38aeb : SUCCESS in 6m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8867d65f9d02414a9fa9344126aefa03 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/daf4b75c359f45efa299a5c9b5b3ddff : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0bc1f13f2f7040e5b10973c63c8f7a99 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de4edbd0059e4e30a480654be43e10cf : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edefb4b3365c499e9aca48cca354ccb1 : FAILURE in 23m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/696aed3e679e479eb26d9dc28ce2357d : FAILURE in 20m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d702ac971b5464080aa25ca4e138c30 : FAILURE in 20m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b89e09d9a8294eb5b8870525feea5b6e : FAILURE in 23m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c96c2b14e799464f829d2b5e121c6a13 : FAILURE in 18m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3faa0a729ad4261beaeb37913acc233 : FAILURE in 47m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9dd63095107d4221a1e1a1d8936a3a26 : FAILURE in 20m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b24ef1e425b3481b8cfaa4b308349012 : FAILURE in 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e90634e781a437a9db31b048828d05f : FAILURE in 20m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9363cece701d4c73831428df14421188 : FAILURE in 19m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/846d35bb7363481198d1087cff318a69 : FAILURE in 20m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0dd862e53d7740129001cc58c9348fa7 : FAILURE in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"949d84d4219f343cbb523f51e9658907c1a7fb97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-28 20:33:17.000000000","message":"Abandoned\n\nI don\u0027t have the time to work on this any more, feel free to pick up if anyone is inclined.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4f00dcf179b6e7ab693a169bf5743b38536cc4d","revisions":{"c3d2fa3628d3c0461b8c5b839b4ab2f7e3d3b20d":{"kind":"REWORK","_number":1,"created":"2020-11-19 23:01:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/763481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/763481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/763481/1"}}},"commit":{"parents":[{"commit":"58bf439bda1cdeaa3dac8b453e05485092bd2174","subject":"Merge \"Fix setting None in tls_versions and tls_ciphers in pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf439bda1cdeaa3dac8b453e05485092bd2174"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 22:59:22.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 22:59:22.000000000","tz":-300},"subject":"Use more inclusive language for amphora roles","message":"Use more inclusive language for amphora roles\n\nAdd PRIMARY as a valid amphora role and start using it\nthroughout the code.\n\nChange-Id: If51065088bd7c5c0189cd1b13f49ef7f6089afb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3d2fa3628d3c0461b8c5b839b4ab2f7e3d3b20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3d2fa3628d3c0461b8c5b839b4ab2f7e3d3b20d"}]},"branch":"refs/heads/master"},"7875a9b37f04c67faf8e6b08bccf65e23bddcf1a":{"kind":"REWORK","_number":2,"created":"2021-01-05 21:30:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/763481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/763481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/763481/2"}}},"commit":{"parents":[{"commit":"4704e458a8f0c5886977aa4c87a6c805a518c43a","subject":"Merge \"Fix periodic image builder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4704e458a8f0c5886977aa4c87a6c805a518c43a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 22:59:22.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-05 21:30:24.000000000","tz":-300},"subject":"Use more inclusive language for amphora roles","message":"Use more inclusive language for amphora roles\n\nAdd PRIMARY as a valid amphora role and start using it\nthroughout the code.\n\nChange-Id: If51065088bd7c5c0189cd1b13f49ef7f6089afb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7875a9b37f04c67faf8e6b08bccf65e23bddcf1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7875a9b37f04c67faf8e6b08bccf65e23bddcf1a"}]},"branch":"refs/heads/master"},"b4f00dcf179b6e7ab693a169bf5743b38536cc4d":{"kind":"REWORK","_number":3,"created":"2021-02-01 23:07:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/81/763481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/763481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/763481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/763481/3"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-19 22:59:22.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-02-01 23:03:56.000000000","tz":-300},"subject":"Use more inclusive language for amphora roles","message":"Use more inclusive language for amphora roles\n\nAdd PRIMARY as a valid amphora role and start using it\nthroughout the code.\n\nChange-Id: If51065088bd7c5c0189cd1b13f49ef7f6089afb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4f00dcf179b6e7ab693a169bf5743b38536cc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4f00dcf179b6e7ab693a169bf5743b38536cc4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
