)]}'
{"id":"openstack%2Foctavia~763732","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I923accd73e0c9cadc91c115157c576432f428622","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I923accd73e0c9cadc91c115157c576432f428622","subject":"Fix load balancers with failed amphora failover","status":"MERGED","created":"2020-11-23 14:12:52.000000000","updated":"2020-12-03 15:56:35.000000000","submitted":"2020-12-03 15:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763732","cherry_pick_of_change":761805,"cherry_pick_of_patch_set":5,"meta_rev_id":"2f1e153174eea17cbe7db1a57a0b70549d1a00de","_number":763732,"virtual_id_number":763732,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 15:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-11-24 10:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-11-23 14:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2020-11-24 17:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-12-02 00:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-12-02 00:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-11-23 14:49:47.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 14:12:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-11-23 16:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 10:05:34.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-24 17:34:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-12-02 00:14:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"35c7054b853684d6a9f1bc1601fd305a7153e2d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-23 14:12:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d35146e75392d380e88463fbc31639ab82b48913","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-11-23 14:48:25.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"6eeb4e8e81cabf5368b160a21332a80a9e13fb6b","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-11-23 14:48:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d8a18a569fdd5dd514b57ac7e57f4589a80d0d6","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-11-23 14:49:47.000000000","message":"Patch Set 1: -Backport-Candidate","accounts_in_message":[],"_revision_number":1},{"id":"efef6fc508d5fc905d62ec0c2715c52d5eeb492a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 16:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1766e0b8dae4f75a500babf83b99d2a : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e467e858e0b941d48259ec5894d491a7 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6634c0cc0035480388bc9997e5b4b485 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e334eff2b943c2bd65f9e1314f6b7e : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4da334a582c54fbe910afea135da7f71 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aeae63a0b14f49af6e37c381acca31 : SUCCESS in 11m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36ea7bc0135443aa921ddd57f2a9d541 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fa234af9b03e46c7aa761ab79ab833ca : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af1c7510de1242a3860144f3100eb7b5 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f810803db349424385396a115692fbf1 : SUCCESS in 1h 15m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10625404a0a941c5adb6531ce52c9809 : SUCCESS in 1h 49m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0bb5be47caa43e8b71c8058f758ccbf : SUCCESS in 41m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fdba7214cb9b4848b1a8ebd277c67071 : SUCCESS in 39m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58cba34f147b4872a66c12ad66db6dc4 : SUCCESS in 39m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2de18871c4f47db8261a6496f03c93b : SUCCESS in 1h 18m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c525bf0c0ee461a8e1f9f3a89f956aa : SUCCESS in 2h 29m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d86e4d927f534394ba4bd07a7ddff31d : FAILURE in 1h 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38554fd4fa5e4940a7c7117d983b72bf : SUCCESS in 1h 58m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33ace38e81e57a5ab0ca8bbf01d602ca9b9cdac9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-24 10:05:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab68a3b5f9ced45760770be557ca6f588440c795","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-24 17:34:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ab74ccc898baab1e3dd529fb0cd359768964258","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:14:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"955a99b139183b9af4712391b1296c85111e0f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:14:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee79d6045cbf6c4a6b78c25f4aff47dcaf08f05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:32:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3e9cc85260f42f788e356d3e5e4718c : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858c5a45af87420eb2edf5132f305e6e : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09a8d8cfc14457291bb8fc70783252c : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/065f078a3d6c4d3faf408fc89fa5463a : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bef7c752c244471862914adc4700747 : SUCCESS in 16m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d6775936d1847a0b4498cd18e28b6fb : SUCCESS in 9m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c17267f46b54359aa671f1e52f34f07 : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeca86d56a2c43f7ba6509ba4e2c107a : FAILURE in 1h 56m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d93cc0249ce34fc0aa433d2331d44e63 : SUCCESS in 53m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8302212d26574a49a86bbb89a69b93b9 : SUCCESS in 47m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b5feba24e3b4025b088af85be4a5fb1 : SUCCESS in 42m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f207520bc1a43ce8dda6ecdf0c1e874 : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"477c244a74ddc2c169d19fd5f896aa349899478e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 07:35:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d539bd738d57a381385ad756ba00886d10e8c7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 10:43:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ff577e0ada04af596d51a01ca2e9315 : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ee3ea5fdb404e31bae03f3525226953 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09fb55f082646c1ba33910430de57fb : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d07cffd6ccb45df96bb97f86f2974f7 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/004250965048444d90f5160fd12fac5f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdefaddb297495a8d764b6a2f428c15 : SUCCESS in 10m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/226075a6c22543fbbe9767a484b32201 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4ec7783827374f259dcad8040f297e25 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c663c84fd8e1434f815b1b372df82fa2 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0326a094e92d4c90994f7646731cd8e8 : SUCCESS in 1h 30m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49a163b3409c42f7bfd4b19f0be7f3df : SUCCESS in 1h 55m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40a494355e3b43819a7d9479febee384 : SUCCESS in 41m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/66beb123efc44f4f802f16a827396682 : SUCCESS in 45m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46b16e89c1b2463f8bdb9ad1af575f34 : FAILURE in 1h 30m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86ee001c0ae649a69231f9d299c54fc3 : SUCCESS in 1h 26m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f076f40e4b244400940f33652ff6c274 : SUCCESS in 1h 53m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2d365ea2d27e4ad2827bf717297cdeb8 : FAILURE in 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c59ff9280ac4f85b27bd4775b7c469a : FAILURE in 2h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1931f37ae5c5c664dc2ad08f58a21b7ee501f4ed","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-02 13:58:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9bfcd93b86baa0e7429c9d25cd8784833745721a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80281057a9c848f7b9255813cd6be9c1 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f475f358664a8da4cd953c77c39393 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1f93bb541f42b888cf48d8425a0b69 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29fc277dc129469d9fc628c42e67fe15 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d7885e388f344cabb96791884f85333 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c856f1565d468885fc00cc12fea1dd : SUCCESS in 11m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9c280500840543a9a0000ff2ee62b2f2 : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e741683ca3e34b0fab35341080ba9a12 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8daea3e2af34fd7a60031eee5da0784 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31e62682d3474541892f91815a8943b7 : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69f543760b754f998d61ae58c3efa2a3 : SUCCESS in 1h 49m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e2ecc3643cc4f90a11271869bf5696f : SUCCESS in 43m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6e4c9eb7a9554416b59ede46942f79d6 : SUCCESS in 51m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/922ad8c4da5d4c5c8d4c7fdcf6637022 : SUCCESS in 44m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc8bda6582064f49bc8e621cc86ac5c4 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a162fb0cbb264b698b8a125c12c8afa5 : SUCCESS in 2h 17m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c81b99748014295900e724f9dff67f3 : FAILURE in 48m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c00dd23351374de69940e4e195e1d5bb : SUCCESS in 1h 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd2c772e595b51bf161439cfb3974e27be42a61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:27:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e24537e7c81ac779968ca2293a27d180aaed858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:14:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07f0eb8de0704235a324450f7e0abf28 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f16cac09731143e98f7b66b2a4d1ecf2 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1fc39c47e3444b584dba35eb2e0724b : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65d81429fecb48a69f9b0832aaf5afe5 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9bfebe9ccf4d38b1b9e5ca66d4e35e : SUCCESS in 13m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92cf7abb43e84c9e89720b43b11f1fc8 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/504981319112429688a8bd55153e3af7 : SUCCESS in 1h 23m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ccb65b3dd6ef4d31b0211866460a698f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/903fbf173e1b4f19b1d09499be67fde5 : FAILURE in 1h 39m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d1d973967334b458f51c68c3ba53bc6 : SUCCESS in 40m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd5abae54bde44629138d99a48ceddb5 : SUCCESS in 44m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e4df8ef238144efc99e72d619ff2b3dd : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6fef3facc848ea1ce087aaa8cbffa51a49a7ad37","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-02 21:19:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c73ed03b1fe133d3a545ce2fe691f975266ba62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:19:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd139c11785430f9da0c623060babe7 : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24d05690289446a5957da8032f98b5f7 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65193ddcff9041b6af41d88cf7ea4ee5 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23c56812b6084b3bada0e37916ffe79a : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e95616edeb8d4a82b8eef31b5e1e3be2 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833472d6c269443e97d1938725f936cf : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e93efe36c2f646809d651fbd25327c12 : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91664d8cf69048f1a7dbabccd3ec6493 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5c70da52ae243aeb14507c547931679 : SUCCESS in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5639e064352f49aa804f317ff3cddbcc : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a7068499c534497bd4e012d7ed132c9 : SUCCESS in 1h 58m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24ea804232b849ada233adb3149a4286 : SUCCESS in 46m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bbeffd8d683f4164b9d406e672598948 : SUCCESS in 42m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4dce0aa49c034cfeb37d1f5d95137e62 : SUCCESS in 37m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af3cbd22a6d441f6b1487bfc6beb4eea : SUCCESS in 1h 18m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b1f6def93b1b4b98ae461df29b175290 : SUCCESS in 1h 58m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73f1618e03e14698be034edbb7e421a0 : FAILURE in 1h 26m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b731827399d64d8297e2441827c4cabb : SUCCESS in 1h 52m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56cdd3b407470f7468a7c250c208759043dafbc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 00:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc1f8a296c530db0c104616c9a9e3de16ba1dcb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 02:49:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffb1f149091945f997ae498ea18a1595 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214ee7a3712443f7ac767c9a46426129 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd32e07de7c24039b05961d0ed6033a7 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4f00ab0419445d191e46d56413af515 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef46dbf4ca0469094ba99bf3a633b3b : SUCCESS in 21m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e99b2a8b26b04394b0720fd798fd6f1d : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddc06111156148f4bae7533da3a6144e : SUCCESS in 1h 29m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae2a6dc091dd41f3b55d878a7a81aaa5 : FAILURE in 1h 41m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0fd92ddd313b47f9813ada69661703b6 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/078ea662016a439bbd569e9603ec8abf : SUCCESS in 52m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/676db02124534d6f982465097879ec19 : SUCCESS in 45m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e13e6381176e4307a7623aeaf17de26b : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"143c7584af570cd153594d159bb5628d4f31caad","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-03 07:05:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"671f64c2b86125f17307dba5ffb10cdd84170642","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 07:33:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44275379ff408c2030987de58be33c873a1387eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 10:41:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d64e5795955846c59fd2e871d459c8d0 : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e044ab936874d14923b7302deeec640 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6f37220f934ac293b017d3f83e891c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d697779c1d014e9889bf0528063e91ba : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5d2b26605ff472a8021bdd14ad4d874 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73831735308f4f30989a752e274a8563 : SUCCESS in 10m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/766efba418224c5c94d7e13afd609cd5 : SUCCESS in 19m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/77f047fb28b9448a8a6a2a7e83b434ee : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cc1251cb73f475d8864b160bae9b0ff : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3e02b58d4d24214a6e9b82e213d77e5 : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e9c30f54a3a47809f30c34faebc3c99 : FAILURE in 1h 43m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9dc3c2773904116a861ede81b538d4d : SUCCESS in 40m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/196d462d847f42c294a3c44a743f95c6 : SUCCESS in 35m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/533dfd31f569452180ede2ec52b11b2d : SUCCESS in 48m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b26f7d76189245b0bb12bdfc3170ca7f : SUCCESS in 2h 07m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfdede2d9152446eb3b4986ceaae3a33 : SUCCESS in 1h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49455308c63b4ad9b8c81b818cb10ceb : FAILURE in 46m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdaab62b24004e50ac5fd4448a87a763 : SUCCESS in 1h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92e8df4358de76ad10fa49b1031c84c7bf8e002b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-03 11:09:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cba5534c2523782689775bbd8547df00fa550e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 13:41:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/060b93f579664e44b1fcc51088ee7937 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c29038fd9674d13b5c35e4b6551084b : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e66ffaa48749d290a010a67ba44fa6 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2634f547b804a1ca5f7f970852b15d5 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/592f5b8f3b2a462ead055da9ff9c8817 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a51b5aa7f7b4410aa00fbdf67770a58 : SUCCESS in 10m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/74220604ce3c4679885ee54aeadb9a68 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6fc44f3b748e476384eddb5c91a24797 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6ee278fa37549e9ae323787f4d02467 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f18716f6d394d3b85d6fbeb2628b6a0 : SUCCESS in 1h 03m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b368d6bea6ac4f30bf9fc0f9daf95f88 : SUCCESS in 1h 58m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d240b3499434b84af2525bf1faff827 : SUCCESS in 42m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f656fca5be49430185578378b5d29da2 : SUCCESS in 39m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2992274e5a8b4a34ab4a70969047e096 : SUCCESS in 37m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d65f192395fa4f3788632fcb39f40489 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c4fe12a21ad4ac6b491be6de005ed7b : SUCCESS in 1h 52m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c93727d00704a62b6e1cb354238a321 : FAILURE in 44m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd9331a5ea904ae09ffdab339e7f10a4 : SUCCESS in 2h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8df734b739cf546308d129e8fdaa67916f29dc54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 13:41:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14da86beb750689bb48c0e8ceb1e788426d6d223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6dc840f514d48beb433447a2914fc0f : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f6b53967e245fb978cdbf39c3ebde3 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d72ca47bcf1141c38fe8ea0e3569b662 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90948d4437b3441ea1444db8e8c6e93e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25da873d5f334f7b9384702996ee59e4 : SUCCESS in 15m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89544ef77fcc4f17b0393a8007d30796 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddf93ef43f73475e80a166703070f96e : SUCCESS in 1h 17m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3208c960be884843b99384c9596b3cec : SUCCESS in 2h 07m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c850e5fd2dc41eda439ff8b2ffca544 : SUCCESS in 52m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1419b6919ff04f4db71b977752a1a28f : SUCCESS in 37m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e76ddcd60ca345a9a0138266cd677b2e : SUCCESS in 55m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a3499c74e1240a6a73a234f0ec1737f : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"719a0583f16ae228d9c2e5c0aefd7534f89f5cac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:53:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f1e153174eea17cbe7db1a57a0b70549d1a00de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:56:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2bd0bd23364601b649dff1de73d2b8 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b31f143200feef214bbdf83a616abdc41dbb9ca8","revisions":{"b31f143200feef214bbdf83a616abdc41dbb9ca8":{"kind":"REWORK","_number":1,"created":"2020-11-23 14:12:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/763732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/763732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/763732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/763732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/763732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/763732/1"}}},"commit":{"parents":[{"commit":"214c386277254131b6174f5822437c0374be1d1f","subject":"Merge \"Remove unnecessary joinedload\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/214c386277254131b6174f5822437c0374be1d1f"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-11-07 00:26:02.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-11-23 14:12:52.000000000","tz":0},"subject":"Fix load balancers with failed amphora failover","message":"Fix load balancers with failed amphora failover\n\nThis patch is a complement of [1] which we ensure\na fresh lb object, but only for AmpListenersUpdate\nclass.\n\nIt was observed, at least on train and ussuri\ndeployments that multiple failed amphoras were still\nhappening when using session persistence.\n\nThat is fixed and failover works flawless with session\npersistence when we also ensure a fresh lb object on\nAmphoraIndexListenerUpdate class.\n\n[1] https://review.opendev.org/#/c/756597/\n\nChange-Id: I923accd73e0c9cadc91c115157c576432f428622\nStory: 2008099\nTask: 40802\n(cherry picked from commit f96251c74275ac44a8ac20ff23a0f6c850af44f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b31f143200feef214bbdf83a616abdc41dbb9ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b31f143200feef214bbdf83a616abdc41dbb9ca8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
