)]}'
{"id":"openstack%2Foctavia~763814","triplet_id":"openstack%2Foctavia~master~Ic294ba786c4493e90fad75382f51f74e9ace179a","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ic294ba786c4493e90fad75382f51f74e9ace179a","subject":"Implement secure RBAC for availability zones profiles","status":"ABANDONED","created":"2020-11-23 17:51:11.000000000","updated":"2021-03-10 16:44:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84af3e5026031bfa1cda43872ea9875a3344ce79","_number":763814,"virtual_id_number":763814,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-24 07:17:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2021-03-02 18:51:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-03-02 18:44:56.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 21:03:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 15:26:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-24 21:57:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"d290a5af7a14f32bb35e294fbb6c0d9ae244e767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715f4cab8814fe4b8c873011dabd29d93fb52c56","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:53:47.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"2f30453007e307ffdc5ec58d7768eaea1345af34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 21:03:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/655c5efc8c8e438786e9cd552cca66d4 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67b346f6307f4135938758c04177f31a : FAILURE in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec74011209c746ba82484e454bcda5a2 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89042fc5dd2c4ecd9226ad4498382171 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86b5b9e5fab74d8385a8eeed7a8c0e45 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c263689131d344f3956c2718fb1bac8d : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cf9ec967da42dba828a5358399d1f7 : SUCCESS in 11m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d3d6c3efd343445c9a2ef75250331e25 : SUCCESS in 4m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/09fee6f9ddc84661880170b7f6fd9b3b : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0db0ea00db04da5b11d9d2fe08984fb : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/145fa054c7cf45e3b206bb52b0e2806f : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c8ef95e472043c4a9e6e94b8ee033d5 : FAILURE in 1h 17m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6fb5925a36ef45c89977f9ca5d7c0555 : SUCCESS in 2h 10m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bed7731928d64a778cd729c8c56e1de9 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2414a73ac2eb46168c925c0b48eddd9c : SUCCESS in 38m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5deaaab015c64d08898b12fbd18dcade : SUCCESS in 43m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/deea4f5c17984214875b2d116eb4f964 : SUCCESS in 44m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1aa95b789e104da7b470c3002679071c : TIMED_OUT in 2h 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46782148146a40cb9c4901ecc33ba7fa : FAILURE in 2h 27m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0906e04c831d4a4782cb1b6b18ce265e : SUCCESS in 1h 46m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7b3257b018f94b538cc1e32497230161 : SUCCESS in 1h 25m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/edbb0fb30dff4c1195b5cc52ab136f9a : SUCCESS in 2h 16m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cf349a24128d4ec29289a22a38e8c181 : FAILURE in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35efff14d81f119f05f2cee5d7cedc844a6dcec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"020e1db88bbb2535026fb9c1ff8deee93b537daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:17:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb1877a95814fbab62f4ee2fd3247b4 : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad779013268c492a94fe099a6cb2112a : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01891d15c1ab46ae905bc14374ecc3e5 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d13c49ae105d4acea32c889048b4bc18 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39a40906a6284fe6ae143e1ba7be83ca : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27c57d6ba4104a6eb76488a654b2144c : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e141fcc3bec74b87903404213d91a83f : SUCCESS in 11m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9aad6f06e65f4da0b26b7f0164bc75c4 : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/291e8048be30482187580388a4047c48 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7fb321cacabe4321a5341db34b0d3127 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87ec61f4b2c04e0896952f22dd3728b7 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aeb451ef09c140719ef9182c62b63cc3 : FAILURE in 1h 12m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac4d543fe3da4c38872dca91209dd9eb : SUCCESS in 1h 51m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb34166548784c68b614528fbd8dfe33 : SUCCESS in 41m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/638a1ff316ee4b4bb0a00ef10319306d : SUCCESS in 40m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e42f5c35eaf49fd9cace22660daea17 : SUCCESS in 40m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4528d42485ff442e88c0ccecdb6e8305 : SUCCESS in 55m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/38b777f9eb69480e826849a8b67156af : SUCCESS in 2h 07m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ddcbcfd22851493ba290ee3dfc431619 : FAILURE in 2h 27m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/abe903e620e3493cac922ad5fb1ad3bd : SUCCESS in 1h 49m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/039110a25e3f4ae58f7a2cb374885af6 : SUCCESS in 2h 00m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ce92d13eadc9469da04a3142e5fb44f6 : SUCCESS in 1h 47m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ae0274bb8ed743048b8f467f8aa5f332 : FAILURE in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0242d65c4ae977b5b9e894e2755ed8d479715816","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:26:12.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3af826ded9b5bd8d1a110499d2c9786d3b77641e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:09:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8e3fc774cba47a9a6bad203fc55f156 : FAILURE in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d80fd30f7e91455a8f5cf02c6ec02f8e : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c37c24fd58a4b07b74c071e90158caa : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bc82803fca64e2aa325b62f7602063f : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca52d10beee405aa37908d7dd5523ac : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2936e833f32a4bd9a664e48c98d948c5 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e260e27aa574789a8cfc986d9179f18 : SUCCESS in 30m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5ee1bc9137b48568985e72d90b34e35 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0f931f197c45465ca00fed766f864acd : FAILURE in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b92fec57abeb4790b771a19fa64b8c43 : SUCCESS in 10m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db20b3fb311d4c84bb461b9f2e07e7a2 : FAILURE in 14m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea76559b5bea430ab773b56d5de40158 : FAILURE in 1h 18m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4327ac95b604771af56d864e7f2235f : SUCCESS in 1h 30m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a70f3d58a3f4486280e5b1bbc28cf73e : SUCCESS in 44m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1dddc4e502594d2fa71ef018fd6dd11e : SUCCESS in 41m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9e8521c6e3e4315859b82f80841b15a : SUCCESS in 41m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78f27ed3228e4ecc8a7e27733575e01c : SUCCESS in 59m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50dcfc82f4404011963b52a942b61be4 : SUCCESS in 2h 04m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d4847145c6a48aca2f7dac48b8b2785 : TIMED_OUT in 2h 32m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2e9c00b2c05400d8ed241297affbf9f : SUCCESS in 1h 52m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f298003623094cfeb375906c48e499c0 : FAILURE in 42m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/611b3a58eedf47889ad096603975e94b : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/04d3c497d18747bbb85631ec36fbb68f : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ffd0da0917654fc84b19e4d22319b88fb5029ef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:57:41.000000000","message":"Patch Set 2:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":2},{"id":"ee548e6c035e0443f58889a7a4e59fdc7b624e76","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:44:56.000000000","message":"Patch Set 2: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":2},{"id":"e0ca7cbd7d7408695d495702695d8dc94f920eeb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 18:51:57.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"84af3e5026031bfa1cda43872ea9875a3344ce79","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:44:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19529a298b9f17e7694707bb904f0760b5518cfd","revisions":{"db4b4c580a6bbdccd7e2894704ebf1243568b60c":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/763814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/763814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/763814/1"}}},"commit":{"parents":[{"commit":"82b3116278f9cdaef79bc346f924132ab7b3932a","subject":"Implement secure RBAC for availability zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/82b3116278f9cdaef79bc346f924132ab7b3932a"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 14:31:32.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 14:50:26.000000000","tz":0},"subject":"Implement secure RBAC for availability zones profiles","message":"Implement secure RBAC for availability zones profiles\n\nThis commit updates the policies for availability zones profile policies to\nunderstand scope checking and account for a read-only role. This is part of a\nbroader series of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: Ic294ba786c4493e90fad75382f51f74e9ace179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db4b4c580a6bbdccd7e2894704ebf1243568b60c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db4b4c580a6bbdccd7e2894704ebf1243568b60c"}]},"branch":"refs/heads/master"},"19529a298b9f17e7694707bb904f0760b5518cfd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/14/763814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/763814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/763814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/763814/2"}}},"commit":{"parents":[{"commit":"0d0cd47c04c0cc5bfb51ccc5699e4bda9a115959","subject":"Implement secure RBAC for availability zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d0cd47c04c0cc5bfb51ccc5699e4bda9a115959"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 14:31:32.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for availability zones profiles","message":"Implement secure RBAC for availability zones profiles\n\nThis commit updates the policies for availability zones profile policies to\nunderstand scope checking and account for a read-only role. This is part of a\nbroader series of changes across OpenStack to provide a consistent RBAC\nexperience and improve security.\n\nChange-Id: Ic294ba786c4493e90fad75382f51f74e9ace179a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19529a298b9f17e7694707bb904f0760b5518cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19529a298b9f17e7694707bb904f0760b5518cfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
