)]}'
{"id":"openstack%2Foctavia~763816","triplet_id":"openstack%2Foctavia~master~Icabcefea9656ebba295048774fa787b8bd7bb9a7","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Icabcefea9656ebba295048774fa787b8bd7bb9a7","subject":"Implement secure RBAC for flavor profiles","status":"ABANDONED","created":"2020-11-23 17:51:11.000000000","updated":"2021-03-10 16:44:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b4ffc3e879a3449d97d361603a554e9608988bc","_number":763816,"virtual_id_number":763816,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-24 06:58:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2021-03-02 18:52:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-03-02 18:45:42.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 20:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 15:27:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-24 21:57:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"6688c3c1fe60325cb2e060858797495c8c7131d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4263bc731c69d43dc4778a57da1b64b05df4a1a2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:54:18.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"a52c7ac72454e60e04bd0f924610428e89d6191a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 20:54:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee376d682de94417821e6abc7049ef04 : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7efe4765f404d7ba79f6624c4d99568 : FAILURE in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f08cdb5a6fa429b8261028f8fa24124 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff827cf81ebf49b9bf9e0def27038ac2 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/332c7cdb7a2046c08e036bdefb2099fd : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fa9a4a7ada14d9d8ee85c234dceac48 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69630c5521e2482a8a2765c384efeeb1 : SUCCESS in 11m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec527a56cce6469f99d6521d6f95e1c1 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3331a1962b242f3861815103699a42a : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e7800ca69bde43eda47f39c3d5b8be7e : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e9d9d86c52b443d9b2bca398b5db11a : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cd34226a2d44fdf89d8319186249757 : FAILURE in 1h 27m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8b289df938d4844a630872254ca2eae : SUCCESS in 1h 49m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8a6d69b8f6b44d7898cb6e1a341a30a : SUCCESS in 40m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d7b4c1b04cb145e5aecf9917eaf3c126 : SUCCESS in 1h 34m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1afaeb4726124b788676bb598ddb0dec : SUCCESS in 43m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/870b296e6aed41efb0e5895a91857581 : FAILURE in 1h 10m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/67f9ab3ce0854255bc630e0e0232dfa6 : SUCCESS in 1h 58m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3bb95fab694c455ab0b5f9ffd5c97cdc : FAILURE in 2h 07m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b58954c2ea7417d98c8a7bb3722971f : SUCCESS in 2h 00m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/17d7742f6526481e86ba08dddc1af966 : SUCCESS in 2h 08m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/414c078a27104d72ac7a1eb40944e6d0 : SUCCESS in 1h 42m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/13a79a61fd1d4851b79c83d19ac4c61e : FAILURE in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"383149407b3c90359f257cef42805ca6e7ab93cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"597f4e190048c1798c402bed3451d6cd0e6d17c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 06:58:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d039f89e2504a27b95cbee90032b8bf : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07c96be9fa394c96b6ef5ebba8c3ce5b : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3ad70ff6d14fd59337d904953bcd7f : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/793ee8b7f03e4caab62c6aedcac0d593 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c20c70957feb499390cf42307390a859 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7bd8e589c0f411494fed1c3c1ba7276 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939af84855fd461ebbeb43e9b256d913 : SUCCESS in 12m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e4386897c984472a64dc24ab54f16e7 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c026ca084b446be91723d8f36ef0412 : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c937bd4ae9e458788c1c618c73e7398 : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec9fce5dd99e48d9bf96059d161d914e : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f27acc801b04d37a4389bc02de53be9 : FAILURE in 1h 34m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e02203fc091c4974b6fe2fee7e0c4c86 : SUCCESS in 1h 52m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ec6693b7ec64ab79717310d64916725 : SUCCESS in 39m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1973fe16e9b647cc95dd757841ff0bac : SUCCESS in 37m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/838d1afbc98d4eb8a50859d3d121a8c6 : SUCCESS in 38m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6af54da96017404eafa6884a62f05c35 : SUCCESS in 55m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/324209e290c645ffb0d45b4960e56732 : SUCCESS in 1h 51m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/883d2f63903a4c2db610bfb885428fa0 : FAILURE in 1h 51m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f408e8ec1ba14529a6a22210febb15cf : SUCCESS in 1h 39m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/085603fcdc824885a0b65649ba23f3ef : SUCCESS in 2h 10m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/672b0877d05945e6bcca2a43a9dbf139 : SUCCESS in 1h 43m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a471d805dd894347bd57efcf9750430b : FAILURE in 29m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7e43410016d11a3569c1c820ab22d7b7ae9b4ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 15:27:48.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8f8ec20a769bea435b5e2bd9970798fe124c43b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 19:42:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb1f3b2213841ee928cd68bfec2803e : FAILURE in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ca73199b06f4706a947416d853e4d81 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9af79e017b94f9d88850c13951119b3 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e0f5fb63cda4cd4917dab64c57614c8 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0d791d907f481ea57821c41cf8b90b : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28161b98b2564692826fc8c8aa531afb : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c242c0cb974254b2ec0397b3cdc105 : SUCCESS in 13m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15fee8c37c774486bbbc70704a9f1f8b : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/823c5829d8ac44df9e3c0dcfe55451a0 : FAILURE in 17m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/92b13144247e442b8ae384ecc5df282d : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91c901c410734cb88608d33f29fc4254 : FAILURE in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef4d29ac8ad84bc7a444fc7f4f3c0467 : FAILURE in 1h 20m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8edcae2b68f847b6b32339d899842da6 : SUCCESS in 2h 03m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ecf1352613b245a08cf17aa319ffc2e2 : SUCCESS in 42m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/afbf558d8c4644f384a1e375e1e2ab35 : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b27410e46dd84316bd49a0cff4d3c8cf : SUCCESS in 39m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3d1c748bb8a45f5bc54751828195464 : SUCCESS in 1h 28m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/baf4e729167c4987b0098f6b7546a1d2 : SUCCESS in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ccd33cbd438d4b0383bf1203eea4076d : FAILURE in 2h 15m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9ed10d2b90e240368a8537f19a3debf9 : SUCCESS in 1h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c03b372aa4e044a68db825041d96ac43 : FAILURE in 19m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d061d759f732481dafd57acc6e3b8c0d : SUCCESS in 1h 47m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/63498d2d575d43d8abf69f5a0ed15f80 : FAILURE in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc01a1b0d9923b6e565997874106f5f8918ea94f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:57:16.000000000","message":"Patch Set 2:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":2},{"id":"9d083d1910c1080eed271c7dde4c88558100a6bf","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:45:42.000000000","message":"Patch Set 2: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":2},{"id":"1dfaa404c1fe0757b0fe78236b61a6181a4189de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 18:52:13.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"3b4ffc3e879a3449d97d361603a554e9608988bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:44:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9784bdec265785a8c85b9439ed323bcfc0667cfe","revisions":{"1cdc0bcb0cf395fcc70f6d31dcb2eb4dfd25a358":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/16/763816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/763816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/763816/1"}}},"commit":{"parents":[{"commit":"b756351999a572ef008e0330e66593420beec89e","subject":"Implement secure RBAC for flavors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b756351999a572ef008e0330e66593420beec89e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 15:54:39.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 15:54:39.000000000","tz":0},"subject":"Implement secure RBAC for flavor profiles","message":"Implement secure RBAC for flavor profiles\n\nThis commit updates the policies for flavor profile policies to understand\nscope checking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: Icabcefea9656ebba295048774fa787b8bd7bb9a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cdc0bcb0cf395fcc70f6d31dcb2eb4dfd25a358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cdc0bcb0cf395fcc70f6d31dcb2eb4dfd25a358"}]},"branch":"refs/heads/master"},"9784bdec265785a8c85b9439ed323bcfc0667cfe":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/16/763816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/763816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/763816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/763816/2"}}},"commit":{"parents":[{"commit":"506a275cd3d6accbdd2c81b3e192b51d6a705372","subject":"Implement secure RBAC for flavors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/506a275cd3d6accbdd2c81b3e192b51d6a705372"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 15:54:39.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for flavor profiles","message":"Implement secure RBAC for flavor profiles\n\nThis commit updates the policies for flavor profile policies to understand\nscope checking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: Icabcefea9656ebba295048774fa787b8bd7bb9a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9784bdec265785a8c85b9439ed323bcfc0667cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9784bdec265785a8c85b9439ed323bcfc0667cfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
