)]}'
{"id":"openstack%2Foctavia~763817","triplet_id":"openstack%2Foctavia~master~I64daa66da432c3a23c0e6ebfc6afbca8cdf10016","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I64daa66da432c3a23c0e6ebfc6afbca8cdf10016","subject":"Implement secure RBAC for health monitors","status":"ABANDONED","created":"2020-11-23 17:51:11.000000000","updated":"2021-03-10 16:44:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7fa2db9c0961fad83e2aad11cc4a4878b07f4ec8","_number":763817,"virtual_id_number":763817,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-24 06:56:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-03-02 18:46:12.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 20:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 21:59:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"ba00b49bfab59c01168f0eeb4b8bee8826d9f105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a046e63beba004fa2d6c263c70f83414ab13709d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:54:36.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"2980b4bf8a8e4d75ac53d241f7ba05a5a4f13365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 20:58:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e0b80556d834684b9ae28b267e670b2 : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d568a6b7c97471288ecac8b195fe188 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74fe11add1c47daaedd6d5fe0dface6 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7554cf2e4b9c4b7a8316cc7855fe4a70 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60778d153f6e4d0e95b4be2c51f0a30b : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8144ccb9c21143feb4c9f39be0de878d : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2736dc5d8e4e3cb1cea46c3d362b57 : SUCCESS in 11m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/79e0c1f514224eb9bd2e792ffe4046e2 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5b16043328864ea0a334eee5a0efefd5 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac3c2762d67d46268d4e7c5bf994020b : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af4d07fd16ae46d1a82b56ad3ccf59d2 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1ba8113708444bba40973f0bef67402 : FAILURE in 1h 19m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67f9dd23d17a4932ac36a40225f151f6 : SUCCESS in 1h 47m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2eaee3bf34e4e6e8e93d95f85ce3422 : SUCCESS in 40m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/464af744d5244659a53899c3414180a0 : SUCCESS in 49m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2b52b307f674e709370039f747474ca : SUCCESS in 53m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4febdc5a37394830811994cacccf292c : SUCCESS in 58m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b3cb73ac079482e98c176ecd4491488 : SUCCESS in 1h 46m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59d887717be4424fb3f90372f63a9bf1 : FAILURE in 2h 12m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f15691281c94b8190a6a5245cee21e1 : SUCCESS in 1h 52m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d83ad66258a1478eb8804f73084125a4 : SUCCESS in 1h 45m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2b0bd23c0fee43e9ba62e15ed919e430 : SUCCESS in 2h 07m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e9ec19d2191b4c6ab8a8dd3fbe7c69b5 : FAILURE in 46m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"545707cd93469e1bdc18ee933de3f3250a88a76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d3c4899beaeab1f86e2c8511eda68593a61f609f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 06:56:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64d6f5670e704a56a0da769a7db3c804 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72140fe9d1664a3486abe8b7328fbc42 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8bb83e0c4984ee0b0fb6dcba6c8ada4 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a375361cd0fb4cec9bb64cb0f4189a4a : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69a26d8c89842a78dab83194e221689 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4140a92ebc614e819cc6babe02ba6fef : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef65a41fe8654ac29652daa958551e99 : SUCCESS in 10m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ae410b8ca92d4b22bcc8f9d5324a1331 : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2d3525b5e3cc4c0a93caea6aa2581b47 : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/459433a25f334efb92337f08ef5d0213 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d5be0774ec04dcc8c865561425da862 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31c6c5003f494c748195c8bcd8cf9bbc : FAILURE in 1h 25m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2076e5915e0e47fcbd24f23f9620eaa7 : SUCCESS in 2h 07m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5ea139dab10455d9250d1175896ef8b : SUCCESS in 39m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/099ff2acf588412aaafac9c2ba2de262 : SUCCESS in 50m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74fe426b0df04a94a42a509f1c353061 : SUCCESS in 44m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a5c07f5c4114fa79c48753c3dfdbde0 : SUCCESS in 51m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/78493a8082104dcc80b213b9cf3bd97c : SUCCESS in 1h 46m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f595479e37964a7fb6dba543bfb03ff5 : FAILURE in 1h 53m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f90dd3e205340338e32e90dab7dfae9 : SUCCESS in 1h 38m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4de9b685fe4143c5a26c17861a8caf40 : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/efcba282cbf04ed18f44651a211fa898 : SUCCESS in 2h 05m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/529972cacebd40f8902c08ddc4b1f5f0 : FAILURE in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d53b68ed94fcceb426bcb68cf554003623dd29fd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:59:55.000000000","message":"Patch Set 2:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":2},{"id":"c39d2b56c7f571e017fe508bb698741a9b6853a2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:46:12.000000000","message":"Patch Set 2: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":2},{"id":"7fa2db9c0961fad83e2aad11cc4a4878b07f4ec8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:44:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"793198296d9c91b27168fb1a0511ca3bee078c8a","revisions":{"16170f0820d4abd82b862859a46618e1977024f2":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/17/763817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/763817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/763817/1"}}},"commit":{"parents":[{"commit":"1cdc0bcb0cf395fcc70f6d31dcb2eb4dfd25a358","subject":"Implement secure RBAC for flavor profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cdc0bcb0cf395fcc70f6d31dcb2eb4dfd25a358"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 16:07:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 16:07:03.000000000","tz":0},"subject":"Implement secure RBAC for health monitors","message":"Implement secure RBAC for health monitors\n\nThis commit updates the policies for health monitors policies to understand\nscope checking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I64daa66da432c3a23c0e6ebfc6afbca8cdf10016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16170f0820d4abd82b862859a46618e1977024f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16170f0820d4abd82b862859a46618e1977024f2"}]},"branch":"refs/heads/master"},"793198296d9c91b27168fb1a0511ca3bee078c8a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/17/763817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/763817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/763817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/763817/2"}}},"commit":{"parents":[{"commit":"9784bdec265785a8c85b9439ed323bcfc0667cfe","subject":"Implement secure RBAC for flavor profiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9784bdec265785a8c85b9439ed323bcfc0667cfe"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 16:07:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for health monitors","message":"Implement secure RBAC for health monitors\n\nThis commit updates the policies for health monitors policies to understand\nscope checking and account for a read-only role. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I64daa66da432c3a23c0e6ebfc6afbca8cdf10016\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/793198296d9c91b27168fb1a0511ca3bee078c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/793198296d9c91b27168fb1a0511ca3bee078c8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
