)]}'
{"id":"openstack%2Foctavia~763824","triplet_id":"openstack%2Foctavia~master~I7d0094e0f0e55384b69b432f0f2a602b785f81ab","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I7d0094e0f0e55384b69b432f0f2a602b785f81ab","subject":"Implement secure RBAC for providers","status":"ABANDONED","created":"2020-11-23 17:51:11.000000000","updated":"2021-03-10 16:44:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2dc24f7401e6639ab3b7ad9e03a01d1382392ce1","_number":763824,"virtual_id_number":763824,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-24 07:36:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-03-02 18:49:00.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 21:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 21:58:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"28ee43e480ad3419072a6f4f696578f13977e959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"496254b8ca40153f996dd08a7789a55f78c1e93e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:55:54.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"5cfd9ed44c4829571ad274d9375410707496d28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 21:48:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ca3cff0c6b646c78d6f2912f4323121 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43219c3c34384841aa5e59071cb147e7 : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84ca4a5ebdf4df59621d61e5eb6c484 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b712d1a9a2c9423f94f7b93d35b4a9c6 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8296a9176b0a4466bb6087a1931f7aea : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/229f444612fa45bea652bdcf06a18b9e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe9120605094ed4b04e252ba65a1241 : SUCCESS in 10m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c2757ef30a0b414d85f9c1766338c552 : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/af03ee0b2bc14445b22f6e2c30c7ba95 : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30067b112df3487da1763b2bc0923a03 : SUCCESS in 4m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00486ab13aad4c39bc408255399415c5 : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5780085773834836b23c4b81823340cf : FAILURE in 1h 43m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/883b075dde3a4ff8908dd9afb5a95ca9 : SUCCESS in 1h 58m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea71302196d04045bf07ec66183ebadf : SUCCESS in 48m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be314aa370414741a8400bf8a482ec8b : SUCCESS in 1h 40m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0220759a7a649598a721a9205289616 : FAILURE in 13m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c60176104934807b6faedb271159a2a : SUCCESS in 57m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c234e911eab468dbda19359ff7b2557 : SUCCESS in 2h 10m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed0e8df2c6284a29bd964da5a84fa4ba : FAILURE in 2h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d4da0fb84eb4df9aaf9c48bb32bbf75 : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72eb4fac58324154b1962fe955ce1418 : SUCCESS in 2h 09m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/138cbb75380644b4b2635b987d504eb8 : SUCCESS in 1h 47m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/47c0564b181c487d9c5188baa500207f : FAILURE in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"142c816398875de44f311760169f380044bf68bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2223a36126d3f308bf625ab9310b2ba80e225169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:36:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38e0889f895e4f99b771053ce36b670b : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1837d44dc4b44532be31ca906d0c9ce1 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f901c81519a94923ae370467b8f146f0 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aad555f49bb4b858db2b8d39d8d95bc : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ffe3b92cc9141f9be73eb284fa637d8 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd131ea1411047c09dbc2115e84cc2d4 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee4d0bd5f35481996c62b6a283b32f7 : SUCCESS in 11m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b3839d82ef8f48c0b653c3d9b6709ba5 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/81e23bb7dac74295b1c7f42472439c33 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/981ca17a084241f6a4845abdd497846b : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e42143aa67c472fa61b0fd94ebd0e78 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b40778f6f8d4a27a98bb96ee4570ab8 : FAILURE in 1h 18m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c89b7fcd0a14645b611b82b633b3ecd : SUCCESS in 1h 58m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a87a2544a58a4ac6acc9b29659f761fa : SUCCESS in 36m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0aaac1141b9940cfb9dd2307609a3e65 : SUCCESS in 1h 02m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30315f35d8d847ec81a87d6109e68e02 : SUCCESS in 38m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d417077f8b0c41b0a75f239ed49b7d73 : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d92648e368c54f3bb3f1a83661912ff4 : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6894bf082a2447a39489878cd67dac06 : FAILURE in 2h 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/28e293b83ee446f790a11424aeda4ff4 : SUCCESS in 1h 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d65d09c675da481d9d7a2ac1dc928ba3 : SUCCESS in 1h 59m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6780d42074b346129aa895b8a420c58d : TIMED_OUT in 2h 35m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cba0d5a6a4c64efcb1c9f194a19172a4 : FAILURE in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de48df21d344605bce4383e35e8a7fbc7c4b6032","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:58:57.000000000","message":"Patch Set 2:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":2},{"id":"7c2314388c56556d1899954576c7e7d9177bfa9c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:49:00.000000000","message":"Patch Set 2: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":2},{"id":"2dc24f7401e6639ab3b7ad9e03a01d1382392ce1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:44:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eceeb64b42f53b34cdfab6aa737b6266610f321d","revisions":{"77fa6c2b1b58c88049dd7097c9a9cd96f8f9ef8e":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/24/763824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/763824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/763824/1"}}},"commit":{"parents":[{"commit":"df2a96bfe622fc0202dcd4d25ebd5f0cb0bff1e8","subject":"Implement secure RBAC for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2a96bfe622fc0202dcd4d25ebd5f0cb0bff1e8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:26:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:26:51.000000000","tz":0},"subject":"Implement secure RBAC for providers","message":"Implement secure RBAC for providers\n\nThis commit updates the policy for providers to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I7d0094e0f0e55384b69b432f0f2a602b785f81ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77fa6c2b1b58c88049dd7097c9a9cd96f8f9ef8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77fa6c2b1b58c88049dd7097c9a9cd96f8f9ef8e"}]},"branch":"refs/heads/master"},"eceeb64b42f53b34cdfab6aa737b6266610f321d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/24/763824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/763824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/763824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/763824/2"}}},"commit":{"parents":[{"commit":"51135997887b19c8bdb075ac57b61e37b190f544","subject":"Implement secure RBAC for pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51135997887b19c8bdb075ac57b61e37b190f544"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:26:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for providers","message":"Implement secure RBAC for providers\n\nThis commit updates the policy for providers to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I7d0094e0f0e55384b69b432f0f2a602b785f81ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eceeb64b42f53b34cdfab6aa737b6266610f321d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eceeb64b42f53b34cdfab6aa737b6266610f321d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
