)]}'
{"id":"openstack%2Foctavia~763827","triplet_id":"openstack%2Foctavia~master~I9dc49fe352f98fa163f353dab8d095c249a4a305","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I9dc49fe352f98fa163f353dab8d095c249a4a305","subject":"Implement secure RBAC for quotas","status":"ABANDONED","created":"2020-11-23 17:51:11.000000000","updated":"2021-03-10 16:45:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4899508536246ab42f4ea4903d50bf6078ef171f","_number":763827,"virtual_id_number":763827,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-24 07:48:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-03-02 18:49:30.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 22:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 21:58:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"3a2f5bcb50d7066159249f6fa2ac9c36ada973d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7532861fae53858ab2f59edf3f2d8389c965f71","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-23 17:56:28.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"08d24999def3c235f5d69a721b5ae789b3acfb1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 22:46:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a538bf3d076e4661ac84afa8b2e85654 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a644bf522534f3d8ffb5e112db71412 : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e9278c25df4c4c8fa12f4ce019f985 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e87c561dc240f08a9fdade884655a0 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55c840b97134eabadaec8b8ec2ab4ef : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a857cca263e4c5588745de8173037fe : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b211882133b741998fa3b0a92718f0b8 : SUCCESS in 10m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8718a33d934d482b992089ada894e598 : SUCCESS in 6m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/985f7ff0ce6644ea8689f97f8ee47093 : SUCCESS in 8m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ba8af3b34cbc440ca77e79e8cc63e258 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4e73279315b4790abdb43ba52bcded8 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/760c0d33cc844733942b80f921a11ade : FAILURE in 1h 25m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d785811493664038adba51b68689d034 : SUCCESS in 1h 51m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7a3c9b3a8e2e48a7b792904b7fb29954 : SUCCESS in 48m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8a94d0c25f147139458daac533764e4 : SUCCESS in 36m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc3afa7b033e4dff94d0ec8b2ec1878a : SUCCESS in 39m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d73120242d24392b0872d8211fa28b5 : SUCCESS in 58m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1f081f0f5cfa4122bac5de0f0ebec3a5 : SUCCESS in 1h 46m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/693949fd7bce42e9bc447b87986a9a0c : FAILURE in 2h 03m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c2cdb28ca84741f89748b9c3488589d1 : SUCCESS in 1h 45m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9a73248a13ab4faeb9fbc790cc4dae2d : SUCCESS in 1h 58m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/12bd124104e14283818d9b20a6cbc2f6 : SUCCESS in 2h 07m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d392e6e1c0654223b89cbb01da66e50b : FAILURE in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3ee60de910611e323cd21e7e2d2215a9c8f11f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"10488196518dba4b53ebb674edc945e1f295f89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:48:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a513d737fff443219feed3cb211994e2 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38e1cc1127564ea09663ecf1e8986561 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1378f456cabc48f2bb2c36ec75232e88 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0f5436e5f7045d69fc879190744d83d : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65c6207e28074a7f9969a34d47e863a8 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f99a6ef562e34fb3a6c02ed13516f287 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ce70de55e147369ed00906d1d867d2 : SUCCESS in 11m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/31427c5a1fb14357a0bc98aa1412d093 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c8c2e83eb61345aea0492bd02b77cb90 : SUCCESS in 9m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f30756e9ad0422a9ce3d5bc83ead808 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dd97fb1ae8f4819b3a3150207b061f1 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/897c0925089c4aa6a457162ceb8709c2 : FAILURE in 1h 20m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce8191f17fd84dd9a5b445a031bc4046 : SUCCESS in 1h 47m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b99a2c4ee3a4fab84666856202000da : SUCCESS in 38m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/50a9493de4384f75b0c71aa8b893861a : SUCCESS in 37m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7da8783b93404eb986ba26245aec844a : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8c09b1c21cb44879dbbf2996caddb2d : SUCCESS in 59m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/daf967d3d9c34a0594f8f98cbcfe5a14 : SUCCESS in 1h 59m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6cd4cf6a4da5487abeb55418e5af5dd1 : FAILURE in 2h 17m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e967495994eb44898560d4d9910a7384 : TIMED_OUT in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aad6244d0fcb4d3188127654e1792f97 : SUCCESS in 1h 24m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f1f6302ecc6a4166aa4fb5587d11566a : SUCCESS in 2h 02m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2d55dd57666044f693d45a00970ffdae : FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8db6aed67f0f3a52942a10f5b5149f4c11205341","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 21:58:17.000000000","message":"Patch Set 2:\n\nAn alternate approach has been proposed in https://review.opendev.org/c/openstack/octavia/+/775957\n\nPlease consider if this is still necessary or a superior approach.","accounts_in_message":[],"_revision_number":2},{"id":"d35a980101dec8ab3dc8440294cf4c854dc15784","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 18:49:30.000000000","message":"Patch Set 2: Workflow-1\n\nMarking as WIP based on Michael\u0027s comment and approach. We can pursue these patches in the future if we want to, but it looks like Michael\u0027s approach delivers the functionality we want.","accounts_in_message":[],"_revision_number":2},{"id":"4899508536246ab42f4ea4903d50bf6078ef171f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-10 16:45:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"974fe988a2d4383c658bf01d45a674770be935a1","revisions":{"a8feefd7431bf19de1f355ebdf2cf91b037769a9":{"kind":"REWORK","_number":1,"created":"2020-11-23 17:51:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/27/763827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/763827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/763827/1"}}},"commit":{"parents":[{"commit":"fe1ea9bcea9a9aff1ec45e1bb3396d14af7d9853","subject":"Implement secure RBAC for provider flavors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe1ea9bcea9a9aff1ec45e1bb3396d14af7d9853"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:45:50.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:45:50.000000000","tz":0},"subject":"Implement secure RBAC for quotas","message":"Implement secure RBAC for quotas\n\nThis commit updates the policies for quotas to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I9dc49fe352f98fa163f353dab8d095c249a4a305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8feefd7431bf19de1f355ebdf2cf91b037769a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8feefd7431bf19de1f355ebdf2cf91b037769a9"}]},"branch":"refs/heads/master"},"974fe988a2d4383c658bf01d45a674770be935a1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/27/763827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/763827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/763827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/763827/2"}}},"commit":{"parents":[{"commit":"b425d7714b4a2a38577c21bb001732fd376770e4","subject":"Implement secure RBAC for provider flavors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b425d7714b4a2a38577c21bb001732fd376770e4"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-23 17:45:50.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:43.000000000","tz":0},"subject":"Implement secure RBAC for quotas","message":"Implement secure RBAC for quotas\n\nThis commit updates the policies for quotas to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I9dc49fe352f98fa163f353dab8d095c249a4a305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/974fe988a2d4383c658bf01d45a674770be935a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/974fe988a2d4383c658bf01d45a674770be935a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
