)]}'
{"id":"openstack%2Foctavia~763926","triplet_id":"openstack%2Foctavia~master~I48e8882f629886d1b8abadff6e60aad91d1169c7","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I48e8882f629886d1b8abadff6e60aad91d1169c7","subject":"Bump oslo.policy version to 2.1.0","status":"MERGED","created":"2020-11-24 04:44:03.000000000","updated":"2021-02-09 01:14:06.000000000","submitted":"2021-02-09 01:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763926-secure-rbac","meta_rev_id":"7624c0578b1a6fbad107673d537bc681463f42ce","_number":763926,"virtual_id_number":763926,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 01:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-09 01:14:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-08 23:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-08 23:20:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 07:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 15:47:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-27 12:22:59.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-28 10:30:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-08 17:22:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"961013d355265da4ad7aac8ffaa9879b7a365b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-24 04:44:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6a683129b498e896388b3a9a9062c5b46e35759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:41:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bab315cc2f784f6ca39ff0e3d8fddc9d : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df23fedb6ad43e4a969bcdd529903ad : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db15619447444035ac493c8d144a651f : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1374f2d5ac4b423a98180fd694368c65 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d955967a3a264a1e81c4514ce43d8620 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa2006048fc40cea31813adff117f23 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06732fae72b048c9921f0b1f9a89db8f : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e67dcefe2e49d5ae6f88c379cd8046 : SUCCESS in 12m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e56cad606faf473e890c42846a22f9d9 : SUCCESS in 5m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f9a1976f57dd4777a35f9da1fda8f1e4 : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/53202b187c5e4ae99f46d8e04d9cd2d3 : SUCCESS in 4m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98745edffb4048e8820e3e092b8653d6 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/103f1601f76c45acaf8cae376789a589 : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9eb3bf2ea75e4cbea8b05abede525e1f : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71058cfc8eeb48f18fb820d4de1d21e5 : SUCCESS in 41m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/15f0b7c9cc88457780b2d00b555b239c : SUCCESS in 38m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59b535d231fc4b94877e1c2c579f974d : SUCCESS in 44m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa174ca8321a44358778a1d9163c31c6 : SUCCESS in 58m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91bb8e21376b4344846e35c907b85839 : SUCCESS in 2h 05m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/acdb4fc49e7146d7b670b2f259e9942d : FAILURE in 2h 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/54e5ed48d17b418389ac808d56c4758f : SUCCESS in 1h 24m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/55a09e4f3e024f74835762cb4f08f161 : SUCCESS in 1h 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/37f8714ab0164f148edb76ec194c0510 : SUCCESS in 2h 01m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fefb2ca3a9ee4304a92b8b6b49f2e624 : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51f2207f708a9ac59369ec6aece99cd106e49f1d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-24 15:47:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c6ee7b857fd9eeab236c2920dcc698ded66293da","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-27 12:22:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"805325f1a8d76a398af17084a4006ab3bad3a21a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:30:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71c43340dcfa1e2909bcac932573e09d92ebf74c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:31:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"968814c1dcf7707ede5cc64210807d4532a808c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:36:33.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d1d5d6fc2e1a0158b7db0b4eba45290a75946803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 16:46:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0436afa77d7694b02dfcf9704362051cdfdc2296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:50:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4178673f8ff419ab93aee7b52e651f8 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/828a79f1c38e4e498f470e03381aa636 : FAILURE in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bff36545d0844c8d86198e08ad8514d9 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796aa9a70c4941eeac5840930460340c : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd120dcd9e204cceaff20ce01aaadd55 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/399deefc4ca94e2b959cca47e31c897c : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/610f8719860c4440829328941c2682a6 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa32f4547fdf4fc580c4c467e22b9e59 : SUCCESS in 12m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9f8a0b1ab7e04a85ac9bcaf644f7ba7b : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f613655ea8c547eaa68c8516d09df457 : FAILURE in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf2eadbc506e438488c3b40bb0b3eb94 : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ccfa643a28b4dbb9720ff38b0f18e2f : FAILURE in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cca1080540f41788fe98a9be62e5690 : SUCCESS in 1h 13m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cba5cb7ed4c74e0a8498eb5e71ead8ad : SUCCESS in 1h 31m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1fd94d63f8a4686bcce899911ead49a : SUCCESS in 55m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/910b1a037365422486e72f269f752400 : SUCCESS in 50m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94ffbca9e7ea403981b74bce4fe2d5ca : FAILURE in 1h 25m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d08eec793e04c6980ec0ab9ed38a32b : SUCCESS in 1h 06m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0eee89ee76754cd197636ed536d33d30 : SUCCESS in 2h 08m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/11e33dec5fe84a43b719e1d0152e8332 : FAILURE in 1h 55m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fc83e08ab38470eafc6eff7fdec848f : FAILURE in 2h 19m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a9d22e32337847bdb2e70935ac2f5028 : FAILURE in 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/108a0b0716494e9ca857e05b0fcf65dc : SUCCESS in 1h 50m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/01377c1246bd44d4b919d83dd63dbc46 : FAILURE in 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8122818a809a05405d2c5db76365d580cc1d229c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 21:54:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c7e8a71d6a4f0981482bd9d7f7b296e0e00a79f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 23:08:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e06923e51019d38a05a590fb5efabc01a6d83639","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 23:09:18.000000000","message":"Patch Set 4:\n\nHad to bump oslo.context to pass l-c job.","accounts_in_message":[],"_revision_number":4},{"id":"9ebc9c80158a198fea8eedbe7ca00ea090b1f09a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 02:12:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71d5d4836a0f4d798c7eb551ba86b81e : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec7528257d8848bc8efd628ffcff8bf3 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b22b7f37ee7c4aa2be3dce24351683e6 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e349d2536644e3cb6db73fa6ec47dc7 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f552a2184d0143f5946346df0b988b94 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679e1c3790534f5195145c558d84b1f1 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9812d70c24345b4be74cf74deee45ba : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8722a4d8fee4e98aad96a645fdd359a : SUCCESS in 11m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9aec04b0e9224b688dbff63bcf724728 : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8f1b3df80e1b42189b710ef7a68f4648 : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1a9357b2a82f4b71ab40d4877fba6563 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/83576027d8944a3b9aa6d697066eb45a : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2242bd6f482b4243a23608483a9107ff : SUCCESS in 1h 20m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e491c1d883524e0a92d86a056c6fdcb4 : SUCCESS in 2h 03m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6a4ad989dc049f3aaf232cdad0b9938 : SUCCESS in 49m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d6e5f35fb9e4214bf75ec0d62ed34f3 : SUCCESS in 35m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffd0b4e1f937437c8cbc542f9f062197 : SUCCESS in 42m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbb7b68008314252b7a83a4876a49420 : SUCCESS in 1h 05m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43a0f3cfd3ec41e593c7b9383e9ab0d3 : TIMED_OUT in 2h 40m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b48a7f1cead43b289d3ff6d77fabe29 : FAILURE in 2h 18m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/326b681058124727bd62f6a5f33bcdae : SUCCESS in 2h 09m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d138992fae3f43df828da78795d84780 : FAILURE in 27m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/da0f301a5a9149ec8770f24b05cf2c08 : SUCCESS in 1h 51m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/738c26c93ee44e189674d11c9df48975 : FAILURE in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d068b381b6e8e50bef6b4a8f85819d3419a8ab91","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-02 11:58:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"04bd8af361016e0904af09cfdd49c8c61c4e6871","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-02 22:12:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f29ff10e238f1d17adf9ebc292a3cb4a0aeb7215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 17:18:55.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"63909eb04dd6ae5501099b42c1f03c386cbf53e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 17:22:56.000000000","message":"Patch Set 5: Code-Review-1\n\n-1 to hold this. There are a number of problems in oslo policy that I think will require us to bump this even farther. Current version is 3.6.2 with more bug fixes to come.","accounts_in_message":[],"_revision_number":5},{"id":"1078eb5e1132aaecc254b68af95782b54a514318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:52:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/672b73d11b9443259c7359bda785c4c7 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b64ad5e2c71f4b97b99e5b7bed0e7193 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2403cc16e544010a1424d0c94de62cf : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24faf2be33aa4072ad353018eb8ab7d2 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc9b1055759475cae1bbf68c7fced73 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bccaa3fbb2084d35861fbd45e4a06896 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78a8598a18404fa1a1b800c3c4b79351 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8417ed54657f42129edd2fb132604a45 : SUCCESS in 12m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bed0729c8163463bb2b9f7dccca15521 : SUCCESS in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e8d1cfb1b6314ffba6188f0f3702879e : SUCCESS in 9m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47cd6c3b649b42cfa31fd56b4fff92b8 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4fa3aadf8c1491ba29f1364a9af80a2 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1616b76bc4944164a0a3e8f7376495d0 : SUCCESS in 1h 13m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ab994d12ade4327b936cebf39dcca74 : SUCCESS in 2h 01m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4af426331d6f416592f427e7a63e0e5f : SUCCESS in 42m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e79bae11c1e940c486ed0f6cbcc6b180 : SUCCESS in 34m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b191761c44a442a7be8a2eccf7dbca3f : SUCCESS in 46m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd63d62375bc432e9ecc9a8a58258dc1 : SUCCESS in 1h 06m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e450cc8da28f4d1d90f869e22dea80c2 : SUCCESS in 1h 41m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ab5cafab2374723b36ba2c88d6214dc : FAILURE in 2h 18m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e2bf482dbee490b8366733132e7060a : SUCCESS in 1h 37m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6aee3a78edc04a6bb31f32201698d181 : SUCCESS in 2h 02m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/abd044028858479dbda8bb6a6d097d15 : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7e2595f95faf49c89959be631a7a2a80 : FAILURE in 30m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6d0e120a2f0978301484edb0c9961b63fc80a05f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-08 23:20:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nOk, we can bump this again in a future patch once oslo policy gets fixed.","accounts_in_message":[],"_revision_number":5},{"id":"546ebf6181994ac2df925c9b2f38f251c4523e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 23:21:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c4a3fdac561997c237ceccbf17ae98b3949203bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 01:12:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f75eee726a7b460ba77787ad3ae26c79 : SUCCESS in 2m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de8fd5f466b442aaa0cdae4503d6bc03 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68cbafd838ec4805bf64e01042a849b3 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/587311e6fce944d38b1097e334fbba2f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97553a1e11f04f958adb7950ab0e29ea : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f3d30429064ddb9adf789e5826fc8c : SUCCESS in 11m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/982e0c6b077c484e9e9376d34b64d9a8 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1fff788e6014c60b6f3774788994b0f : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6c1058c87fb4002a8751e4d8f63e267 : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9d0fdfe708e42ee9ca99805bb34418b : SUCCESS in 1h 49m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0372c7ce25a49389c028f99e1d29f4c : SUCCESS in 42m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/929f6d964517458c9c5ae2f786014a90 : SUCCESS in 41m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9921de6294dc4b94b534fb97cb7602c2 : SUCCESS in 41m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de454cd8bbe547d8aa41ebbc9beabee3 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":5},{"id":"dad8989f51ce96753fe60aa12d9c208ec337c3bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 01:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"00ee5b3b90478a22e394b12fc554557c03a77e2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 01:14:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764128fa32fd433d8f2f748eb400a5e5 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a822f30eb135b31f1e5fb212810f99f5d5adf477","revisions":{"e9ae51610e1d6b659942b926dcb49ad15f30f530":{"kind":"REWORK","_number":1,"created":"2020-11-24 04:44:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/26/763926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/763926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/763926/1"}}},"commit":{"parents":[{"commit":"8302148018213b389e0d84384d802fb2af18ea89","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8302148018213b389e0d84384d802fb2af18ea89"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:38:43.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:40:38.000000000","tz":0},"subject":"Bump oslo.policy version to 2.1.0","message":"Bump oslo.policy version to 2.1.0\n\nThe secure RBAC work requires the of oslo.policy\u0027s scope_types argument,\nwhich was made available in 1.32.0. This commit updates to version 2.1.0\nso that we\u0027re using something more relevant.\n\nChange-Id: I48e8882f629886d1b8abadff6e60aad91d1169c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ae51610e1d6b659942b926dcb49ad15f30f530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ae51610e1d6b659942b926dcb49ad15f30f530"}]},"branch":"refs/heads/master"},"f1833a49d78db96bb085220be3f565fe31707c95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-28 16:46:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/763926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/763926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/763926/2"}}},"commit":{"parents":[{"commit":"8793b93900e747f5f46c4768d1c278fcb4f73610","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8793b93900e747f5f46c4768d1c278fcb4f73610"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:38:43.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-28 16:46:03.000000000","tz":0},"subject":"Bump oslo.policy version to 2.1.0","message":"Bump oslo.policy version to 2.1.0\n\nThe secure RBAC work requires the of oslo.policy\u0027s scope_types argument,\nwhich was made available in 1.32.0. This commit updates to version 2.1.0\nso that we\u0027re using something more relevant.\n\nChange-Id: I48e8882f629886d1b8abadff6e60aad91d1169c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1833a49d78db96bb085220be3f565fe31707c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1833a49d78db96bb085220be3f565fe31707c95"}]},"branch":"refs/heads/master","description":"Rebase"},"cfcb0ecfc2af46c7fd485d0b3345929963efd1be":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-28 21:54:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/763926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/763926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/763926/3"}}},"commit":{"parents":[{"commit":"7fbea7cb1ad336e6a84d27bcc588b961cdffd36f","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fbea7cb1ad336e6a84d27bcc588b961cdffd36f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:38:43.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-28 21:54:54.000000000","tz":0},"subject":"Bump oslo.policy version to 2.1.0","message":"Bump oslo.policy version to 2.1.0\n\nThe secure RBAC work requires the of oslo.policy\u0027s scope_types argument,\nwhich was made available in 1.32.0. This commit updates to version 2.1.0\nso that we\u0027re using something more relevant.\n\nChange-Id: I48e8882f629886d1b8abadff6e60aad91d1169c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfcb0ecfc2af46c7fd485d0b3345929963efd1be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfcb0ecfc2af46c7fd485d0b3345929963efd1be"}]},"branch":"refs/heads/master","description":"Rebase"},"c6dc0878a661d6199a92ed14ecfefc0adf58ab07":{"kind":"REWORK","_number":4,"created":"2021-01-28 23:08:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/763926/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/763926/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/763926/4"}}},"commit":{"parents":[{"commit":"7fbea7cb1ad336e6a84d27bcc588b961cdffd36f","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7fbea7cb1ad336e6a84d27bcc588b961cdffd36f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:38:43.000000000","tz":0},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-28 23:07:57.000000000","tz":-300},"subject":"Bump oslo.policy version to 2.1.0","message":"Bump oslo.policy version to 2.1.0\n\nThe secure RBAC work requires the of oslo.policy\u0027s scope_types argument,\nwhich was made available in 1.32.0. This commit updates to version 2.1.0\nso that we\u0027re using something more relevant.\n\nRequired oslo.context bump.\n\nChange-Id: I48e8882f629886d1b8abadff6e60aad91d1169c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6dc0878a661d6199a92ed14ecfefc0adf58ab07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6dc0878a661d6199a92ed14ecfefc0adf58ab07"}]},"branch":"refs/heads/master"},"a822f30eb135b31f1e5fb212810f99f5d5adf477":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-08 17:18:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/763926/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/763926/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/763926/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/763926/5"}}},"commit":{"parents":[{"commit":"7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8b0db3c854d3c452f6072e1a96c3ac5008c5c7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-24 04:38:43.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-08 17:18:55.000000000","tz":0},"subject":"Bump oslo.policy version to 2.1.0","message":"Bump oslo.policy version to 2.1.0\n\nThe secure RBAC work requires the of oslo.policy\u0027s scope_types argument,\nwhich was made available in 1.32.0. This commit updates to version 2.1.0\nso that we\u0027re using something more relevant.\n\nRequired oslo.context bump.\n\nChange-Id: I48e8882f629886d1b8abadff6e60aad91d1169c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
