)]}'
{"id":"openstack%2Foctavia~764448","triplet_id":"openstack%2Foctavia~master~If94b59e00282e8077711945917a208317767f78a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"If94b59e00282e8077711945917a208317767f78a","subject":"Fix FD limit warning message in amphora","status":"NEW","created":"2020-11-27 11:37:07.000000000","updated":"2020-12-01 22:49:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"d3609dc247f166d40899d901801cac8002af5b42","_number":764448,"virtual_id_number":764448,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-01 22:49:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2020-12-01 20:00:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 14:17:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 15:18:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"3949c7334b42ed971a2b4b05419fc3b17034f07d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-27 11:37:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"853007e10ea022ddab26b924a128c1a85e0cf8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 14:17:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66f6104e1e4145c68957409252faaeaa : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce76b6faf354287af545d840cf04a8d : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55e7cede168477aba004d0dd2b27748 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d810774a6ba4f32b554f24bd2590ebb : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e844915c08284862ab12774233bad655 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bb67795d8d24f0d961ec9ac9247c96c : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f3aa6482a04c709eda3fab10b7b97c : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a25d442dcb3f47ab919f9e958da2d578 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bdb7fde33a3a4be28664cae8874aaeda : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aa6d71bad7534fce8db7ffb8da201f56 : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bf84128cc414a1186ce814c0a179a7b : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07a8cbe2f5364d2a98408ee319fe34f6 : SUCCESS in 1h 13m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b28ebc360ce8493e9669eebe5d57a4ef : SUCCESS in 1h 57m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8553b048663f4701b48b4a49459173ae : SUCCESS in 43m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ccdab0a93d944398511117bf630804a : SUCCESS in 59m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad941a5adfcc446ebabcd82aec12d2d0 : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2f035839fd543899c525b327d5b5825 : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fb1de724f65471b96738b07c9b72efd : SUCCESS in 1h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58a55ea888184c0b81d31665931adbc8 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d3a0c2e99d84ac3a5aab7274d6baf75 : SUCCESS in 2h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/41acbefc6a9e4775ade82c7bf7d9ad56 : SUCCESS in 2h 16m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/31f849dd2cb647fe87cdf3690a1f04dc : SUCCESS in 1h 47m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a6bdc9ffa7514651bdfe38b80aca3698 : FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"669ec5ff3f199febea49d8344f5b2c5860cfd8c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-30 15:18:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41c6b464ebbe39393b126c12865b7a5bed9f243e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-30 17:04:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"44ff75e8ece2f57a853ae2b675e7f4e1bb1f9be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:23:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a8b2d783160468e8ed215c9c4f6ea6b : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25f80bccf3034a86a988b1c07a172cce : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93028edee85e494bbad0f87dbdc78073 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36eadba5a73e4765aa5ab55e1a49267d : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b1690017b7b49c784fff6baa96ce01d : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aa301aab5564dc087c68e0b96c00491 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791c510aa4f24cc2ab2f86074af8b9e7 : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d910817ccd1d49659233a38877f75d00 : SUCCESS in 4m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1fb13fc573824cfcafbb8187cbcc1c0a : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3a31b30a9684ff09f3e6b05b61a81ca : SUCCESS in 4m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d96a1cf821974d20b8ea86bd3da90637 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f1184530b4249abaf2c29db4cf7ba84 : SUCCESS in 1h 16m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/728181aca3304d4eb94906b71501673c : FAILURE in 11m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c60a31bc00640b1a802b26890e06b0d : FAILURE in 11m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/affac23868674e21b8a02a6942979cfb : FAILURE in 9m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0b3196bacb84ac8ba39bd8c1efaedb3 : FAILURE in 11m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c543a6c9d3a4cb2bf69f1973f18485d : FAILURE in 9m 26s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3b6380df202949e8bf872e54621ea304 : FAILURE in 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa259dc5480a41a79978019a805ebde4 : FAILURE in 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dea68fbb37b84ce5915fd5d88a06b94d : FAILURE in 8m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4fa3efaa48ab43959a64bddda0e9f45d : FAILURE in 10m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e7453538166b48bcae3519058397ae36 : FAILURE in 11m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6b167006bfb24fb4b30cd993d05c34ab : FAILURE in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f9434ded7c87a02fda92c1c17c73d105edf10d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-30 21:55:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"829a993257233620251f016ddd237b2eff102874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-01 14:58:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"669020edc2e13f70364bfd9b68154056ff11f207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 18:06:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/844ffa18375b4af2b66d008d79623d58 : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5ef81bf1f7842c9978b4f88e312df9f : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9afba8e3d66b47c9954f04cc44fe3a05 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e455938d7b3a456c93bbd7d87e3cc138 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc6e14a87e844eb9962488ac394d42fc : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ab20432039a4c78841f800d4e14846b : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5a392499b046679803c0115a368442 : SUCCESS in 11m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/18a4311174254e6daa2c955e01b0b0cf : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ba9a263f97fd4283a6a5321b346ed118 : SUCCESS in 7m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eab01a8750844ee6adecacb982a86a02 : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b616d2e07c54442d9d590b91340e9d08 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8cc24ffb36894042ba399921c474e273 : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ccc22ce39f0459b889e00f051d3e4a0 : SUCCESS in 2h 04m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/08981234ebb44e7ab3408972837df5e6 : SUCCESS in 42m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5abca472cc82433d9036cb8e1bc164b3 : SUCCESS in 39m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/715fdbf6ba9143979c9e9e6e70c885a2 : SUCCESS in 36m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9a37f171f494e9c9caca61fe7582ad3 : SUCCESS in 1h 16m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6668dcce47c5468ea164198ad7cdc704 : SUCCESS in 1h 56m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d1cdac5e3d34b41b7ad054f57205a2d : FAILURE in 2h 10m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50f8d530d4934248b580455d6196ebde : SUCCESS in 1h 39m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/92326edfaa114d02a2f0b3f2d28e9bcf : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/53a617e443804ab1b21b8eb28887f5a9 : SUCCESS in 2h 11m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d8123b5e1c2748a3a81306017d9dd9f0 : FAILURE in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"da7e3489fcb0b2e499840ad05c8a4b47fa3a85ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 19:59:39.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b79072088101ebb9f871351d26abca271a675ac2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 20:00:20.000000000","message":"Patch Set 4: Code-Review+1\n\nJust updated commit message to have task #","accounts_in_message":[],"_revision_number":4},{"id":"d3609dc247f166d40899d901801cac8002af5b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:49:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adab4391fff4494eba0808570d25caef : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e234bed97e7c463c9295b004d5d7b3c2 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68faa48201d1429d980936c0c03b26ff : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/497537453df943e0bd033b9690924c88 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb9281d34dd4186b9fc4e60d4343f11 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b263658d304d40caa227117d0564a562 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79ee96dbc37406bad515b90db479958 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a52deaaca1043a485d084b1ef18b135 : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c9b4d22e28724558842e9987625d75cc : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd891c69ef49400ebc6f4581adb2907a : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6f7d9481aa3469b9c335a9ed81808d5 : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ccb17fbf08a4315aeb4ecb530736b99 : SUCCESS in 1h 20m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c74b1883b7ba4948968096916fdb5138 : SUCCESS in 2h 17m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e184d2dd30d548e0b8328c98702675b8 : SUCCESS in 39m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/646600f3a9ac4d759177810d4eaf4df2 : SUCCESS in 39m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2301eb44f01c4d9787b9412c941b4dbc : SUCCESS in 40m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/079b0b6a7c2643eeadaf769b6210676c : SUCCESS in 1h 04m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b930d37d24d34fe292a6f32c30194887 : SUCCESS in 2h 11m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27ce5945cbd5499891d5cf13549ea878 : FAILURE in 2h 28m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca16dc6ab64a44c781947fd13d0f9181 : SUCCESS in 1h 59m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0366724c6ff94a0c9042d32d7c97cb05 : SUCCESS in 1h 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/23822ba1ff10416ca4d2a9470b7e2e7e : SUCCESS in 2h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51702d388cb048b493b8b7a9d56d230f : FAILURE in 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"71b6e34074d096e00f80cb0ba1bb4ef3b609594b","revisions":{"cbb2938b7ef6c5c2f7f686926cbb71a366b37ea9":{"kind":"REWORK","_number":1,"created":"2020-11-27 11:37:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/764448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/764448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/764448/1"}}},"commit":{"parents":[{"commit":"e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073","subject":"Merge \"Fix missing cronie package in RHEL-based image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-27 11:36:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-27 11:37:03.000000000","tz":120},"subject":"Fix FD limit warning message in amphora","message":"Fix FD limit warning message in amphora\n\nWhen loadbalancer has unlimited connection limit, amphora image will\nclaim on insufficient file limit. So during haproxy reload you can find\nthe following warning message in log [1]\n\nWe increase this limit to avoid this issue and make reload process clean\n\n[1] http://paste.openstack.org/show/800494/\n\nChange-Id: If94b59e00282e8077711945917a208317767f78a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbb2938b7ef6c5c2f7f686926cbb71a366b37ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbb2938b7ef6c5c2f7f686926cbb71a366b37ea9"}]},"branch":"refs/heads/master"},"8cf414b279f9329742235a822d3ded22d9a07b37":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-30 17:04:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/764448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/764448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/764448/2"}}},"commit":{"parents":[{"commit":"e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073","subject":"Merge \"Fix missing cronie package in RHEL-based image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-27 11:36:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-11-30 17:04:07.000000000","tz":0},"subject":"Fix FD limit warning message in amphora","message":"Fix FD limit warning message in amphora\n\nWhen loadbalancer has max connection limit, haproxy during reload\nwould claim on insufficent limits for it\u0027s operation.\n\nWe increase this limit to avoid this issue and make reload process clean\n\nCloses-Bug: #2008399\nChange-Id: If94b59e00282e8077711945917a208317767f78a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cf414b279f9329742235a822d3ded22d9a07b37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cf414b279f9329742235a822d3ded22d9a07b37"}]},"branch":"refs/heads/master","description":"Edit commit message"},"346e95e2d1f76c1382e955e2ab84bf692a44b306":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-01 14:58:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/764448/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/764448/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/764448/3"}}},"commit":{"parents":[{"commit":"e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073","subject":"Merge \"Fix missing cronie package in RHEL-based image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-27 11:36:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-01 14:58:10.000000000","tz":0},"subject":"Fix FD limit warning message in amphora","message":"Fix FD limit warning message in amphora\n\nWhen loadbalancer has max connection limit, haproxy during reload\nwould claim on insufficent limits for it\u0027s operation.\n\nWe increase this limit to avoid this issue and make reload process clean\n\nStory: 2008399\nChange-Id: If94b59e00282e8077711945917a208317767f78a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/346e95e2d1f76c1382e955e2ab84bf692a44b306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/346e95e2d1f76c1382e955e2ab84bf692a44b306"}]},"branch":"refs/heads/master","description":"Edit commit message"},"71b6e34074d096e00f80cb0ba1bb4ef3b609594b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-12-01 19:59:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/48/764448/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/764448/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/764448/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/764448/4"}}},"commit":{"parents":[{"commit":"e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073","subject":"Merge \"Fix missing cronie package in RHEL-based image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9e8f5229d9c1f86e5ff5c9d993057a1cffd0073"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-27 11:36:42.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-12-01 19:59:39.000000000","tz":0},"subject":"Fix FD limit warning message in amphora","message":"Fix FD limit warning message in amphora\n\nWhen loadbalancer has max connection limit, haproxy during reload\nwould claim on insufficent limits for it\u0027s operation.\n\nWe increase this limit to avoid this issue and make reload process clean\n\nStory: 2008399\nTask: 41332\nChange-Id: If94b59e00282e8077711945917a208317767f78a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71b6e34074d096e00f80cb0ba1bb4ef3b609594b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71b6e34074d096e00f80cb0ba1bb4ef3b609594b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
