)]}'
{"id":"openstack%2Foctavia~764888","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","project":"openstack/octavia","branch":"stable/ussuri","topic":"redhat-cronie-stable/ussuri","hashtags":[],"change_id":"I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","subject":"Fix missing cronie package in RHEL-based image builds","status":"MERGED","created":"2020-12-01 15:40:55.000000000","updated":"2020-12-03 20:36:18.000000000","submitted":"2020-12-03 20:34:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764888-redhat-cronie-stable/ussuri","cherry_pick_of_change":762916,"cherry_pick_of_patch_set":3,"meta_rev_id":"a32dc7708e35fdd22f423209b0a86e3731813d9b","_number":764888,"virtual_id_number":764888,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 20:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-12-03 13:59:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-03 13:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-12-03 13:59:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 18:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:48:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 13:59:05.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"f5e6e8dc5c22e5b402f835ad57d1741d4221af68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-01 15:40:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fc65feef0f0690f0c8467780770ee3d2a1c74d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 18:40:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9f4b7df9b74493da18cae40b26e4d09 : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d64105e486b74a7abb5c90ab4c76bc06 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0cd864c45fb4764a6025b9b4b3be70b : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cf20c49214e4943b09994877b8f3dab : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d61758b1836e4772ba5b3a1c1372a417 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b6ef69bf6d46619b548b7e493a6c48 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/312ac53f8af84c3d8f1286f20fcb5ee5 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/87229a95a01549a2a2218e54bdc10315 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/777a21ff93f84feda8702331052afe9f : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab6b07e21b924e46afad5a3c8175d8f0 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cc94cd9b7ab491fa32506703c463686 : SUCCESS in 1h 27m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de13d4db0a114323851f573c203a4478 : SUCCESS in 1h 47m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f4d6667ad6f4544a9b4ef5087ebca5a : SUCCESS in 48m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/40b6bb3250df4cbebaefea10b2238ad9 : SUCCESS in 42m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cee8f304a737477a802f6246a51b2754 : SUCCESS in 42m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9dcedcc8b68043fa8b2f1f4b5136f339 : SUCCESS in 1h 18m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/551eb7062662459b99eeebd913b0f5b7 : SUCCESS in 1h 48m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdb4ed48c9364dcfa4bc0e84b9d9b1aa : FAILURE in 49m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a9ab12046df48f4a8aeb103bd47814e : SUCCESS in 1h 52m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ebae3d0da6a4880ea98a78f548fc9edb0f8f874","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:48:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df1be30f7feb582dcaf8c1bc34afc32a3275e2b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 13:59:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42bbe02445c245ba5b696a7d0c74a0df43bc4d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 14:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b567514dc0f17052a04382243cb601b0d141da4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:34:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/147bf258d26f4e0883243f667ffbb6a6 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4468cfb06b4816b2488d9570681fc7 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fef3500f33841d9b5e4a14516d81d2e : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40941f7d7c474144a5957e25c95db1bd : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa3f02b7590437488da57468820ca52 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9771cd0baeed4a90b5f370dcc4c5bc96 : SUCCESS in 20m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fbc5470cf1e84b2e84b25c6cd5795b8f : SUCCESS in 13m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6683a75b0b8d47dcbd044904434063f8 : SUCCESS in 1h 25m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de6479f366444f4b92d672db6983ef63 : SUCCESS in 1h 40m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2a1ff93691846e2b504868e45147973 : SUCCESS in 41m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/16f747ce3d31484cbb2d414a33e39a5e : SUCCESS in 39m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb844500675e4524ab22580773bb35d7 : SUCCESS in 48m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcb9e9bf2911425fb38a35ec7062da65 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3971ae0ae58a7d2b983da33ff7717a2bd91fcbd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:34:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a32dc7708e35fdd22f423209b0a86e3731813d9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:36:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dccc12c25c7742689c76da4a50ec6cee : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa8598038c9b4b18826b4587f49144f1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12846655241b04534e8e363ec79a01c814194b83","revisions":{"12846655241b04534e8e363ec79a01c814194b83":{"kind":"REWORK","_number":1,"created":"2020-12-01 15:40:55.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/88/764888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/764888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/764888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/764888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/764888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/764888/1"}}},"commit":{"parents":[{"commit":"214c386277254131b6174f5822437c0374be1d1f","subject":"Merge \"Remove unnecessary joinedload\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/214c386277254131b6174f5822437c0374be1d1f"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 03:11:41.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","date":"2020-12-01 15:40:55.000000000","tz":0},"subject":"Fix missing cronie package in RHEL-based image builds","message":"Fix missing cronie package in RHEL-based image builds\n\nIn RHEL-based distros, cron is provided by the cronie package. This package is\nmissing from the amphora built images.\n\nThis patch maps the cron package to cronie for distros in the Red Hat family.\n\nChange-Id: I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3\nStory: 2008351\nTask: 41253\n(cherry picked from commit 6aa23358eb2f4df412e010fd1fbb9e35ee2e1522)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12846655241b04534e8e363ec79a01c814194b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12846655241b04534e8e363ec79a01c814194b83"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
