)]}'
{"id":"openstack%2Foctavia~764889","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","project":"openstack/octavia","branch":"stable/train","topic":"redhat-cronie-stable/train","hashtags":[],"change_id":"I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","subject":"Fix missing cronie package in RHEL-based image builds","status":"MERGED","created":"2020-12-01 15:41:59.000000000","updated":"2020-12-03 15:57:26.000000000","submitted":"2020-12-03 15:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764889-redhat-cronie-stable/train","cherry_pick_of_change":762916,"cherry_pick_of_patch_set":3,"meta_rev_id":"0802f70a274855133a14b5626e441aaa1337a6cb","_number":764889,"virtual_id_number":764889,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 15:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-12-03 13:50:58.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-03 13:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-12-03 13:50:58.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 19:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:48:39.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 13:50:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"e317750adf52d31735193e020a087ce789f23fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-01 15:41:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6617cdf8fa7c861f543ce45cfb4567dbd5f14b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:18:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/823c4eefb35c4ecaa4d474f440b94cbf : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0920c1f26d30421b80d4b8625da63a87 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5964c17eb9d4c7fbc2b9b6f5858b1a3 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b95da5d486e4b4aa6aaeb8704d6584e : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f31c0c726f4b489ea5398e6b306dba02 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2efb5fc2e62d46f7a722e38b8bfb8b06 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f089ec69e4844e089105c8652f3a0337 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3db636f36064a3b81668595fd5e9462 : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c22bf4497c9d41acadbb7bd7c01ef2dc : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8739ac4c238c42e19131501489a60b54 : SUCCESS in 7m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/984d8b5a6b5843a6857b912f590abbdf : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8115d656544c4ca5945dc94536166005 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a020eb21a9f4d4e903eae0cff00dac8 : SUCCESS in 1h 18m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/798568ff2c044a0ca7960ecb91b9d965 : SUCCESS in 1h 34m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25a386d8f97f4532b22b15bc52936501 : FAILURE in 1h 54m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fd91ce3de2eb470f8a6e44a22fab4fc8 : SUCCESS in 1h 57m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6e3a3ee08784487098772d39d98fbd5f : FAILURE in 11m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2115a5e6ea60482eb332df7c2ffc8581 : FAILURE in 27m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/56313d096c84464ab84d8dd651efac5a : FAILURE in 1h 14m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/545939837b4445108376ce49d0b6c343 : FAILURE in 1h 29m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8bd54c81248747c0a851247cf88d6c9a : SUCCESS in 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f0e4598e5da414a84c6c1e3d162e5fc : SUCCESS in 1h 51m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/674e4c6e74084734a4e726509f5bc409 : SUCCESS in 44m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1ef2ef34bd7945838722b5e9226a14ab : SUCCESS in 40m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cd8c01feb23b40daab321a724c610b4f : FAILURE in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e38af2031bb602f60784ca51f64cc9d5548d214d","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-02 02:12:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cb0835ffc91bfd019413f87b6a32505a509306a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:48:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c019410e46547c49110f10f8c47bb67 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dde28451fcad4b53bbb565652f3a6960 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a369779aed4911b609b3473ee5a38b : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d97cb52c0fb048dbbff422f64b837228 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b3c41686d24d228a150202a23a103c : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42130bd2c954450d957c690a48fe60c2 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75bfbf9a54a461ea51f8162da1822ce : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/141b9c7c78ce40e28e8f039ec7b4ac49 : SUCCESS in 4m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a7b69ec10894343869cfddaf09c9ea6 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ef0d18740deb4f01a2aa68972ad1d12c : SUCCESS in 7m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/99826d500e8645ba9165f12b94c176fa : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2197bcece4034f20b16554e577084592 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/25e8bf4c7d304e6f9172f7bff9c644f2 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6d57d3ac31824908ae20e9586af77729 : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66bf1e64c05742ef894051d2a4552faa : SUCCESS in 2h 01m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b2cfae19cde4476aaf98f36db3d94dbc : SUCCESS in 1h 56m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/be2ebf7a5bf246fc87b4badab0ef3063 : FAILURE in 12m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/989152976e2b4d47abcb0adc7cefaf9c : FAILURE in 26m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c50cbad70de84b038999c2a8a263438d : FAILURE in 44m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e598bc24480f46c4bd0946b38881bee1 : SUCCESS in 1h 26m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f4cae961b00440ea35129037012c21f : SUCCESS in 40m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1d1707858529484aaed07fdb45a77788 : SUCCESS in 1h 52m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2336b05236d7428eb3b58daac13f6874 : SUCCESS in 40m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/958582fba63a4f72b91f70f6d7798033 : SUCCESS in 37m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fd995df437b94dddaa9004588e234cab : FAILURE in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a1a6fba94c38f3104bd89ce8f14815f78fcbcd5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:48:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edfe36ce100699141391a10d04f8b552be10614e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 13:50:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac84ed8edf02dc2c10029f510925a7daa80cd25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 13:52:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3687de8aeb73148e5f7461c071a9e4a22751307a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:54:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98e49e7ff74c43e0835d6d8ca7f09928 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5d25c1bae84de5875bf2eeee14a7f2 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b6fd753ee114581ba815f78834aa2d4 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b3d5451dfbe461b8e419b2ffcf26ee4 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b69289c81d3a4da58b3448b097b9ba5b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdaf631f53047ed8062433ae308c710 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed99541d20694446b2004f30d0f548e4 : SUCCESS in 4m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/97f650c54e8b4b59b79f029046c7a575 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5f38cda9f8c468dad2cf51ac8ed5c50 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5df387c2ec3249cf8e659ddd9311bdc9 : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fe9d638d121442f282ee9be474b8d4ab : SUCCESS in 1h 21m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/258a09f2889444cd9c5e72cb6c71290b : SUCCESS in 1h 36m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f154957daf045a99a8283d2aeaad3cd : SUCCESS in 1h 39m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0f10f9056bc9639712d34a8b3591d7272e0ef740","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:54:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0802f70a274855133a14b5626e441aaa1337a6cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 15:57:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2afd9a40f3248bcaf58971200855ce5 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1053c426e2e4b0ba943ab9dc02f4d8b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8a999fea41825afeacf3a87126b12ba72acab55","revisions":{"c8a999fea41825afeacf3a87126b12ba72acab55":{"kind":"REWORK","_number":1,"created":"2020-12-01 15:41:59.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/89/764889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/764889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/764889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/764889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/764889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/764889/1"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 03:11:41.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","date":"2020-12-01 15:41:59.000000000","tz":0},"subject":"Fix missing cronie package in RHEL-based image builds","message":"Fix missing cronie package in RHEL-based image builds\n\nIn RHEL-based distros, cron is provided by the cronie package. This package is\nmissing from the amphora built images.\n\nThis patch maps the cron package to cronie for distros in the Red Hat family.\n\nChange-Id: I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3\nStory: 2008351\nTask: 41253\n(cherry picked from commit 6aa23358eb2f4df412e010fd1fbb9e35ee2e1522)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8a999fea41825afeacf3a87126b12ba72acab55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8a999fea41825afeacf3a87126b12ba72acab55"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
