)]}'
{"id":"openstack%2Foctavia~764891","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I3c830a9f0fd21ca10238946721c0f75d01d3ddff","project":"openstack/octavia","branch":"stable/victoria","topic":"redhat-cloud-utils-growpart-stable/victoria","hashtags":[],"change_id":"I3c830a9f0fd21ca10238946721c0f75d01d3ddff","subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","status":"MERGED","created":"2020-12-01 15:44:33.000000000","updated":"2020-12-03 23:12:32.000000000","submitted":"2020-12-03 23:11:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764891-redhat-cloud-utils-growpart-stable/victoria","cherry_pick_of_change":762928,"cherry_pick_of_patch_set":3,"meta_rev_id":"f85ee8e73491ed540b98edc51f9784e5dc14c970","_number":764891,"virtual_id_number":764891,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 23:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-12-03 13:59:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-03 13:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-12-03 13:59:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 20:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:48:03.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 13:59:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"8ecfe94c565f08da9937f36be3333c4d6f816489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-01 15:44:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3c81d11c6d94c0e58c5b2fbe883228baaf56a282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:14:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7686f04b0f9412b99b77cb7e6ed34e4 : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3651f65d14214f5d9ddf2e5eddbde0e4 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e05d36211c114bae8a4e681ba0e0734a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd1b9aa129b44e5dac14086029ce0e9e : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f719f2818e224092a830d7841ce5a1e2 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9c30b3895946eebd064a9efcdc4ce7 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c80dd8c714a4c2c93a0c6ce71383f45 : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eefbfa6e117d43cf9a95e331d89745b2 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31c90cb8d7394348b8a19fc4ab54c752 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ffc6586e88924f2b8e931f008405dbd9 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63fd7e4e3ace483f99b8dbf471ab5e6d : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b2127951d2d44c4958a22a5a465dc39 : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c6d42f4242841a7b651c5e9a7415e75 : FAILURE in 2h 16m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e754810fe70142b1be181259ce61eda6 : SUCCESS in 49m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc407b390abc4c73a09ed7b03daa9329 : SUCCESS in 48m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4bd3f18273541e7ab336b1ce9ed405d : SUCCESS in 43m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bea5fb9e89c4f93906e84a23b024fbf : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ec74056f5564c60b7e44c01b6bfe271 : FAILURE in 1h 44m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b8006cc9f0c458ab20def2d274eb0a2 : FAILURE in 2h 02m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/57ec9a92130a4ef8a8ced06376ec5703 : SUCCESS in 1h 48m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f311fd48cb9849d8ba8894b40d718d20 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/12e1e18f9c2b44f29791a9cd40a1e3e1 : SUCCESS in 1h 49m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/788ea42cc7234dda84440f740e7dfee9 : SUCCESS in 1h 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3857af9bf3e0a05b93ae788872de365696838e50","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-02 02:09:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2be80b0370088846614575395d5ba58196e49ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:54:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e91fb524aabb41dcb0075eca3372562c : SUCCESS in 14m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d5dc3ab90fa4aa4b7f1cbaf48b2731a : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123e92be29e84ce39bac210b93598476 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33add731260d4cdeabe850c15ae96923 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a33849dfaeb4f5b9df676e450a02be9 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c7c481dba440beae47257afb53f67d : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/384af7ccda0a40759c70d75d091d6aa4 : SUCCESS in 4m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4755ed69f47c4aed8091e842b4a6f610 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1138d86e3217480baa8957fcba99faf2 : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f922dfe57dbe48959832e48cd764923e : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aba5acd11984439f98c4ea5b4d86c05e : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64b11fd51866470d97c00137ce19b723 : SUCCESS in 1h 29m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40180ac82cc0495c82461ee82290207c : TIMED_OUT in 2h 35m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f703b3fd0084aa28f27c4db79092704 : SUCCESS in 48m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1aee4aea295a4414871ae6cfa02ff833 : SUCCESS in 44m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/192ab21f26ab4868b990161bbf9c8fae : SUCCESS in 42m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8901dd4ff3724355adb9f818931186b4 : SUCCESS in 1h 03m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d9b4c8b98b94f438e75af508911c025 : SUCCESS in 2h 19m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/027366b9dd724cde97f45fb3dc6df3f0 : FAILURE in 2h 29m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8feb0eee584fe0b0a874b359c8c91e : SUCCESS in 1h 48m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e8aa00e4b514cc8ac6d29a8b3b56e37 : TIMED_OUT in 2h 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cff3291c47fb4b7183dfd513e86a78dc : TIMED_OUT in 2h 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4c3cc704f7fb4756befba9fd0dd23ab6 : SUCCESS in 2h 03m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15f03936c88dcea76f0bef3d66ab2bbd91f83fbc","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-02 18:26:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1601dc8416e9cc2422c5facf0d8a082d53f82de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 21:06:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd2f80fff0d4b42baaa4394e4f999d3 : SUCCESS in 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c449251ade33462592bea93307baa14c : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2fdf7642244c87ab6708be9a8d80cc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3900a0be8e3247229e126c4a8861d9ac : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4df5a3c3fef4fbdabf10f8251821f68 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ffda9c1a2b41c7b28323a12632f606 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42a8dd98801c4d83829274b1873d4b6b : SUCCESS in 4m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d42f4c31368746d78e522cbf61c1ec30 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f4a4c76c19d9445bbb97878d249b3355 : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/381a522173934182a2ce763b2e24b20e : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6905b3a3124a402cad74dfcda5034c40 : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e184d6c06654b82a31adfb7550c68d5 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af896a57639a4fd794f0ec099f689055 : SUCCESS in 2h 00m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8e67845d053e4269ba8b2fb94ff25564 : SUCCESS in 42m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c4ab9453371149a2b8c21ddc271d8afb : SUCCESS in 38m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6a95a94409a46d1a5101176375403cb : SUCCESS in 50m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5752edc53d334f8f9d66e59d1fd766d6 : FAILURE in 1h 29m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0becc6baec7943628598f7c3d36061b1 : SUCCESS in 1h 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3fc1b65f4f2643a5a7488eac1dcded22 : FAILURE in 2h 15m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be4d647aa046490bb98c4baeb161b0c1 : FAILURE in 2h 21m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b35ee2333c744465adbfd4890d391bf3 : SUCCESS in 2h 01m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c57660177a474c529a6259c09bf53122 : SUCCESS in 1h 46m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c74db695d5d64c2fa1c819743c62781b : SUCCESS in 2h 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2b82513a3c601ca866f703c025b1058ca24cde6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:48:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a70a081c78f53b0259c976d339475e308d3d3b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 13:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"490c7775dc4568516abcefe796bd256603d36987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:24:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8099e670f1bc40c5a99aeb6ed18fcb3a : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/198af46ef2a9424ea2e81ad07e4c0340 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edae694f579143ae9092408c79d0ae69 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40e25ed6da1e4b7b9f53e59411d41ecb : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba9693ae656c4681ad23159697f230c5 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973f588a12a94d76953c3a5b918c2d55 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06e65937d2a94643a17c43839410b541 : SUCCESS in 5m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ef5fd9699d847b89bf1d65ee10b20fa : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0e9fb4d624b848ce90cde8d4f5d4150b : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/479657c06c414edf85cc2216ece1cdf8 : SUCCESS in 4m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4153b18fd8194d79bcad43b52b1160c5 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f7a0b58d99924dbb84da0ec021e11903 : SUCCESS in 1h 18m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8be67884697a41628efe475ad8111f7d : SUCCESS in 1h 46m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8745dfa4efc546f2b076f4d3cee0024a : SUCCESS in 43m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6dfe70746096400e9eb3adab6f7b9539 : SUCCESS in 38m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c372d723ec654179b9fc325a9dabba0a : SUCCESS in 42m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/881afddd9cac4b33af1e3aa62971a77b : SUCCESS in 1h 06m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15dcbb7a08bb491d865cba4ff80c0224 : SUCCESS in 2h 12m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b620bef520564e35b196c654052081af : TIMED_OUT in 2h 32m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9bedab770425469599d2bf35a5a9fbf8 : SUCCESS in 1h 53m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8f92bb3b4aa496f80c0eceae866a0f3 : FAILURE in 2h 08m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4ab8f54a870e4c9087081e13dcae14a8 : SUCCESS in 1h 57m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7098f198e7b347748f38fd998897cb84 : SUCCESS in 1h 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2843517c6b7fd7772d50093d720ebed2c6cb58f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:25:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaae4dbe5dbf0344160cd650464ef75eb3d95d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 23:11:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c58e9febe2c7463b81a33e52dcd9de07 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2421db62fa4b39b5793c86806a3a4b : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f2a29d2fbc84a1593a811e3c417d3cc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b260f6fa15a4d10b026dfd109fc358e : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e436c879a4fc438f84b8351318d6ae77 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0db18ddf5b104ea28ef37852dd02245f : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ca28c77ef5e42a8b60bb6cedcaf3b65 : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e499a58171fc42299a88999fef365698 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/680041fb432e4b5c97d0c72efdab82bc : SUCCESS in 1h 17m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/242b5d2384164e91934a53c2ac4a9f22 : SUCCESS in 1h 34m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ab490daa05649e1874bd948052c7498 : SUCCESS in 1h 11m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/edca71dba25147de95c29ddad4806b06 : SUCCESS in 48m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea230184107540af9b2d478c4246182a : SUCCESS in 45m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/164146a134074d94885b5ca21a955726 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c0b4d3ac3c9efda497e8793e2392087f6ad5e267","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 23:11:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f85ee8e73491ed540b98edc51f9784e5dc14c970","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 23:12:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02033ddc4c2f4196b55312dead81efb2 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e877b38670d147a8bdbef2c144aabd5a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"895238230a903c4eca6b1897526e5069d434d0c6","revisions":{"895238230a903c4eca6b1897526e5069d434d0c6":{"kind":"REWORK","_number":1,"created":"2020-12-01 15:44:33.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/91/764891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/764891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/764891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/764891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/764891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/764891/1"}}},"commit":{"parents":[{"commit":"fc058c3a8367c369ab623083c34cb423f808c52e","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc058c3a8367c369ab623083c34cb423f808c52e"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","date":"2020-12-01 15:44:33.000000000","tz":0},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based image\nbuilds which means the root filesystems are not grown on instance\nstartup which means not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n(cherry picked from commit 6d59fec8a0c8011949c547c4f49195c41b9d21a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/895238230a903c4eca6b1897526e5069d434d0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/895238230a903c4eca6b1897526e5069d434d0c6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
