)]}'
{"id":"openstack%2Foctavia~764893","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I3c830a9f0fd21ca10238946721c0f75d01d3ddff","project":"openstack/octavia","branch":"stable/train","topic":"redhat-cloud-utils-growpart-stable/train","hashtags":[],"change_id":"I3c830a9f0fd21ca10238946721c0f75d01d3ddff","subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","status":"MERGED","created":"2020-12-01 15:46:11.000000000","updated":"2020-12-03 18:28:54.000000000","submitted":"2020-12-03 18:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764893-redhat-cloud-utils-growpart-stable/train","cherry_pick_of_change":762928,"cherry_pick_of_patch_set":3,"meta_rev_id":"9fb261cf39f4d712c253a4ca7cd8900f893d567e","_number":764893,"virtual_id_number":764893,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 18:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-12-03 13:58:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-03 13:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-12-03 13:58:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 20:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:48:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 13:58:17.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"d2140df0318e723e30ef2b41bb9b09960c985048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-01 15:46:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"220e59444110eb58d9d64271ebcaa4503b09d6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:42:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95454d28b2934e3ab7f8b90090cf8e84 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8823f47e34134e7f980c8b99c3d6cc12 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95906fe6ea24bd28ad0564a3443a974 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a622ba7079584b78910bbe4d2e5c6468 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abb25d46f69a4b6abc238e280151879e : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bbf9bdf352c40c092324dc0de3796de : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7e38b3b3bc4b8c8dc6346bab4af382 : POST_FAILURE in 13m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be6ce3367a8a4e0e9471f3a6089970b0 : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/65a98a91d96e4d979b19020a10ceb93f : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e4e1596d8a564220a42c14a94698a247 : SUCCESS in 12m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f4db4a35e9b54f68a67bc583e8ccef90 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea52ff6312114b019b830c6347dd267b : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8c8716dc17a4f1a83ac3926630cde11 : SUCCESS in 1h 18m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6dba028a688044d5b9abda7ea14cebff : SUCCESS in 1h 32m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4628fe54abd411c8891aaabd6b623fa : SUCCESS in 2h 13m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/665e178e40814483947f9af502dd7ea4 : TIMED_OUT in 2h 34m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fa91b6ccdef94a2680b841d40460e8af : FAILURE in 11m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/590e2d6766084909a79d8e1002e1748c : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d1e0912d610442180730dbcf2f1f995 : FAILURE in 47m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee87633aeab24cfaaae09c4673916a15 : FAILURE in 1h 21m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4fcf8a3777f42e096bb0d86751e63d3 : SUCCESS in 52m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bea609d1198f413e81f09ebb1091fdd5 : SUCCESS in 2h 00m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d641576f8a28401bb3d0c86560aed8bf : SUCCESS in 42m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a2a351ecf8464f2e9eee29881048e201 : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/96919f2b175b40af95e6681b019c3142 : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3d73cd9241eb3edad0d32ae32280d8b63fe4ea1","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-02 02:11:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83a827f4352b5afc78ee941322a20faeab60d819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 05:00:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9b1e5c1ae164c15b2f616bd97dbd3e0 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1f0e0662cf743ec9939a8c10759b85e : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0134f8ceccb44d88b34d8abe7533e5eb : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb116314b5634db5bb8ae925e2d278a1 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2875c8d453c34fd5989d708936a77960 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e4275bacc0d482eac6332ea2ad2f534 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec1da51852043ac821be6189b73dbc4 : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/993e903504324e368a068b602f7af7b0 : SUCCESS in 4m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/88069674fc8f48de932469f677b50f8f : SUCCESS in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2c0268755a643a5a0688d839ba19458 : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a9f5a5828239435095bc2307224be15d : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0275858658514d658c91cf37d6c0342d : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37bd89ed8b8a4f63b11ff029fc594260 : SUCCESS in 1h 14m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ee7dfe4b89b24c3d8b1c7472b7906f2d : SUCCESS in 1h 17m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8c31674d93448ed97f6d430d4dc6ab2 : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57fb88b82d0e49c1990cdb96442e1079 : FAILURE in 1h 48m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/263f1dd90e3c4d91bbcc619114b935b7 : FAILURE in 11m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66b68a720a764a3195f82ea981937923 : FAILURE in 27m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c8126451d4144579e5fccbf0f572628 : FAILURE in 52m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57614ce41fbc4545a956445508f61e09 : SUCCESS in 1h 26m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ba3b32fe81e40c9b3f27c06aaaaa4f5 : SUCCESS in 57m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/300f4896b13e453eaf336def63f44fc5 : SUCCESS in 2h 21m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a992181109524f5aa2de1b5e2321a528 : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4371ade0f76e46928751820b499d4486 : SUCCESS in 59m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7e2961ce05e544429ce50c774e87efb1 : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77f6abc538e84cd191b082727472dcaccd52d0ef","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:48:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6d7fa0a17f75bd2931fda96bda0ad34925f087c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 13:58:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e122757a4dcd2315846d7841f068a19fad562d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 13:58:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6ea2bc46aa70920b4cb36860bb02e21a4cdf2b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:26:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9475009ff1b44f2a08c0df823b903cb : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ed4141c31c403eb71b5b02261ee7b8 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fac9a18c8a0642b78efeece92e2b8e4d : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b2e1dbd30a14f0aaea3295a8106d6e5 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aec0de76041d4f6885e1feec796ed448 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe77710da61419794faeb4c426bc561 : SUCCESS in 12m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d9a420434b64b4abbfaced32a656938 : SUCCESS in 5m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/308ba48e98a1449cb0a49d411ea542c2 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/251b751564b74faeb2ec9c2782815a8c : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0fdbebdc9dd643b5bddb060ecfcbda58 : SUCCESS in 1h 14m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/33258a61a3d249cc8cb95d18988d213b : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6412bdd58d0496ca3e26d13707d9443 : SUCCESS in 2h 07m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99c7655ccec44d2da5f78874356a582d : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6f57ace91ad1bccad2b033ae57be125e4708adc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:26:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9fb261cf39f4d712c253a4ca7cd8900f893d567e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 18:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d7186d73224142aefa1bbcc730e510 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd799e515e648c3816b91496747a767 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"011f1b05857527639b68b0115d32dd41fe15ed93","revisions":{"011f1b05857527639b68b0115d32dd41fe15ed93":{"kind":"REWORK","_number":1,"created":"2020-12-01 15:46:11.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/93/764893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/764893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/764893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/764893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/764893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/764893/1"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","date":"2020-12-01 15:46:11.000000000","tz":0},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based image\nbuilds which means the root filesystems are not grown on instance\nstartup which means not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n(cherry picked from commit 6d59fec8a0c8011949c547c4f49195c41b9d21a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/011f1b05857527639b68b0115d32dd41fe15ed93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/011f1b05857527639b68b0115d32dd41fe15ed93"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
