)]}'
{"id":"openstack%2Foctavia~764894","triplet_id":"openstack%2Foctavia~stable%2Fstein~I3c830a9f0fd21ca10238946721c0f75d01d3ddff","project":"openstack/octavia","branch":"stable/stein","topic":"redhat-cloud-utils-growpart-stable/stein","hashtags":[],"change_id":"I3c830a9f0fd21ca10238946721c0f75d01d3ddff","subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","status":"MERGED","created":"2020-12-01 15:46:43.000000000","updated":"2020-12-03 17:45:54.000000000","submitted":"2020-12-03 17:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764894-redhat-cloud-utils-growpart-stable/stein","cherry_pick_of_change":762928,"cherry_pick_of_patch_set":3,"meta_rev_id":"b32db883acd54ee04d4bbfb7991af94cd21d64d8","_number":764894,"virtual_id_number":764894,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 17:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-12-03 13:58:08.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-03 13:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-12-03 13:58:08.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 20:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:48:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-03 13:58:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"2bfab9e0d04daa8ce60845b96316360618d32e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-12-01 15:46:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"54e00a2287c9a5d0098843a77a31553c4ca0cdb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:05:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a97abea00d4dd987a36a3d0b4ade7b : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1523a80fbbd4428d8a0ab30596292d2e : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0044b527b7347e7a4f7f979506f97be : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/181c11edd0994f28b008f45f3b97c7a8 : SUCCESS in 7m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b2dbc015c164b28b52ba5df7580cc71 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4158138cfb404a82b6e4b449488da0a0 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cf842f491a42e08fd5d756bd18fce0 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87783973541b44458cd46c22a617409e : SUCCESS in 7m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/fc5f23e106ba4d0ba40820ae322cc0b5 : SUCCESS in 1h 16m 42s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/4429b980370d4348b515320f1362ea6f : SUCCESS in 1h 26m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47de59ecab6247aebe22795e64a08329 : SUCCESS in 1h 37m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/09b5c9355f4647a3a73862cbc2954bf3 : SUCCESS in 1h 48m 37s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb6127fab6264294bfeb478663f7886c : SUCCESS in 7m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/62cdb865ef684d52b1076d7330c97649 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/925add92a8b447499243a865d0183f1d : SUCCESS in 7m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c41b827cf031499a8cfe3a71f80dcdbf : SUCCESS in 55m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f768dec8b50648f2ac4c5918b16e3783 : SUCCESS in 53m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/557ee39361c24983a483189506c18d33 : SUCCESS in 1h 12m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d879fd22a07449e89f8f4f76f657d9a7 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb35797de2fa4d7d8fa27f10082c1794 : SUCCESS in 1h 17m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0ca7797007ff4026a616ab95cf7c328d : SUCCESS in 1h 34m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7e477cc1c14f482f9d45154c61e99bb5 : FAILURE in 11m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adf2a8a8e5944a1b9b544d2809e42906 : FAILURE in 52m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c7f16704b04423a98af70fb6a1c316c : FAILURE in 46m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dba35dfa9eaa43adb683abd3ad382b5a : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a178cabc8e28481a8cfedce771ab67d5 : SUCCESS in 43m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d0222d551d73806aa7583c0a815d25f23fb6a5f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:48:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"692a34eec142c5a50c62d7f0838a79fcd35c97d1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 13:58:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ce9cc0d8c485bb3e031953cd07e77f45e7cee02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 13:58:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdcfad34f6b4898cd951113ada7446344e83b475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:43:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2692c7277fca41f1bb7735b35930833b : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f0897b53d743098171c92759283e1f : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68fbfac6ae1e4e8e849ad3a615a136a4 : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8949155e7f8240c89e58bb995e40d01e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c20c3cfa95547d6a19a6c89ce144842 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb14e67370b4db49380cf9bc1106a42 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d30e2df2232142cdbb1e8f625b59d7c1 : SUCCESS in 5m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16a44ad78613414aa6e44d89a402d019 : SUCCESS in 6m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d8f4246e64984d36b10c4e63696ce550 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/918fb8efd56f40e2898adf253c555b50 : SUCCESS in 7m 48s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7df0cbfaf8a4d3ea1bd2897f013cbaf : SUCCESS in 55m 56s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b928b7c679564b0a830781c104b8416f : SUCCESS in 1h 07m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/272959d8f39641aba2e88c471839e659 : SUCCESS in 1h 14m 36s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/fce150c5ef594e4495248cb9cd29a7e1 : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e08df48c3409465890ade655771f716e : SUCCESS in 1h 07m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8449396efd454a6eb9bcc625bcbd70b2 : SUCCESS in 1h 04m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9c06a091e1f445c8242afb9415e378a : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f72fd66d553000449aeb2c9f4507bea14d536d31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b32db883acd54ee04d4bbfb7991af94cd21d64d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:45:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f56e46b54144ff990e6ef4f50a74dc7 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20586f2f5e1846e78d99a6ce322f5cb5 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1425fd53fce0d0ca6d0abb6082ac0f3fb2b8853d","revisions":{"1425fd53fce0d0ca6d0abb6082ac0f3fb2b8853d":{"kind":"REWORK","_number":1,"created":"2020-12-01 15:46:43.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/94/764894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/764894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/764894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/764894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/764894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/764894/1"}}},"commit":{"parents":[{"commit":"7eb216605a61f16cfaca8e71384cd7b733bf99dd","subject":"Merge \"Remove unnecessary joinedload\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7eb216605a61f16cfaca8e71384cd7b733bf99dd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 18:43:53.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","date":"2020-12-01 15:46:43.000000000","tz":0},"subject":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.","message":"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\n\nThe package cloud-utils-growpart is missing from Red Hat based image\nbuilds which means the root filesystems are not grown on instance\nstartup which means not much space for things like logs.\n\nThis patch maps the cloud-guest-utils package to cloud-utils-growpart.\n\nChange-Id: I3c830a9f0fd21ca10238946721c0f75d01d3ddff\nStory: 2008350\nTask: 41252\n(cherry picked from commit 6d59fec8a0c8011949c547c4f49195c41b9d21a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1425fd53fce0d0ca6d0abb6082ac0f3fb2b8853d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1425fd53fce0d0ca6d0abb6082ac0f3fb2b8853d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
