)]}'
{"id":"openstack%2Foctavia~765240","triplet_id":"openstack%2Foctavia~master~Ibb461802bedbff9a49af339d0c66e853709e16aa","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ibb461802bedbff9a49af339d0c66e853709e16aa","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"ABANDONED","created":"2020-12-03 05:48:56.000000000","updated":"2021-03-05 05:56:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"82b755c0391b892babe95ffa596315b96264541b","_number":765240,"virtual_id_number":765240,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-03 20:15:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2020-12-03 16:40:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 16:39:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6d0a3644104e6e15f329cd23ca78daee075bc889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-03 05:48:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aab0889b1910d2da5d7ac480d81bdde280e99e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 08:26:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30f1e46de9ee46dc8f3aefad7ae6f3fe : SUCCESS in 20m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22bafb142a8b41a58cc6e930057907ec : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bebc808ae8947a699dd17ed88153b2e : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888d2548a9ed4272a62f4644fa2348eb : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b314b8cfdaa847c381f8d8eabca79a4b : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/567b1c80b8644790ad021124e19a55f2 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef422b8d122345268ada28b44e1bdabf : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a9295484544b39bb73b413e1f6112c : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc54d73f35134c869bf9040158345882 : SUCCESS in 22m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/22232f1931504685b0492c043912d0e0 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e0da8a97ebcf44eeac95be87f8d7c156 : SUCCESS in 6m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1955fbf76c46447b840d4df62c2bb16f : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/342d0d4bbf7e41c5b03ff056c9ab1cb8 : SUCCESS in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a65ad10bb7241e18e394c72c37435b2 : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86594123cfe64f10a41d9c0aba0aa014 : TIMED_OUT in 2h 33m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2097d65683c4d6499ef775590a1bcca : SUCCESS in 1h 07m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1537f0f3dd6e41a8be7472bfc98ef08c : SUCCESS in 35m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b45d45f66d844f7d832956ac8529bc34 : FAILURE in 2h 08m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/be80ad05937749afa79cfbdeff78e0fa : SUCCESS in 2h 18m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/50e063dbfba5481f87ba52ee4dc8b1c3 : TIMED_OUT in 2h 32m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/11c9f4130a17444a8bae3a26ef332b1c : SUCCESS in 1h 52m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb32f2e486d14646ae2fd2aa1554f35e : SUCCESS in 1h 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/479b84a2db134cdbb298cf76de5309c0 : SUCCESS in 1h 48m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ccf8a1c8c9034b87bd1991deede3221c : FAILURE in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d81ac13b7ac8c0dbcf87d857689a32e594811a5d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-03 16:39:42.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ddb11c77486b4f5b914c92221e80f472c168a9f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-03 16:40:48.000000000","message":"Patch Set 1: Code-Review-1\n\nActually, there is already an ongoing change for this:\n\nhttps://review.opendev.org/c/openstack/octavia/+/760711","accounts_in_message":[],"_revision_number":1},{"id":"cedb7df9b0edd7b123e9645a0cb21892f5a7629c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:15:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/77912b1af16c42da9b225a3449e22edb : SUCCESS in 21m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faf0cd0ed26c4406b99366423a7ba197 : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fea9262b98f24bae8881d5a208f71804 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f45192bff554d98829d81b689e1be73 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/448d11b7ca9d4df283ca92029b8a3b4e : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddffd323d5894e7e9f10dae98217062b : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2f5395113a748b3adec13032e05ef61 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf078c56847f4aeba8771796a9ba2b05 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff3a13ebde624732b70b333a927a6bab : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f06d999f07494c499fdde54735971ffe : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5592fe0a9f9843b6b6eaa42cacfaf10e : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6de71f2d239840fe9198b88b1eae5433 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa2252a9101c43c38084d2f08fce4d55 : SUCCESS in 13m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea9d7daa2d72464c8081fe4c9e71e3e2 : SUCCESS in 1h 53m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20fd088aaeec4c91ab5419b97d4d1e01 : SUCCESS in 2h 13m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/263a5163222942e98ee89e186bf0b417 : SUCCESS in 37m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ccf691a3d9a146e787a5c8739216917c : SUCCESS in 36m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a49f1333704a4e3d84cfed5abec91cca : SUCCESS in 41m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3fe364144b44c6588b70b7db69b0fe7 : TIMED_OUT in 2h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b25475476994584abfb511fcb288dfd : FAILURE in 2h 15m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be8a9f4a93f34f589417365ae51d5382 : SUCCESS in 1h 39m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6bfe47c963304e5eb5287a7f456ae381 : SUCCESS in 1h 45m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ca8c241d39304cb78d19902543280382 : SUCCESS in 2h 08m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/30f86a6aa76645b39b2c5cf8e65f9afd : FAILURE in 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82b755c0391b892babe95ffa596315b96264541b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-03-05 05:56:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21eab27d6679f3c9f69b210d667a3b93b5ba88a5","revisions":{"21eab27d6679f3c9f69b210d667a3b93b5ba88a5":{"kind":"REWORK","_number":1,"created":"2020-12-03 05:48:56.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/40/765240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/40/765240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/765240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/765240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/40/765240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/40/765240/1"}}},"commit":{"parents":[{"commit":"8258992c01f0a35112b4aaa3dd1bdec31a9368e2","subject":"Merge \"Fix nf_conntrack_buckets sysctl in Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8258992c01f0a35112b4aaa3dd1bdec31a9368e2"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-03 05:44:45.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-03 05:48:50.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nChange-Id: Ibb461802bedbff9a49af339d0c66e853709e16aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21eab27d6679f3c9f69b210d667a3b93b5ba88a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21eab27d6679f3c9f69b210d667a3b93b5ba88a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
