)]}'
{"id":"openstack%2Foctavia~766381","triplet_id":"openstack%2Foctavia~master~Ib089affa229531cd72f6853105d74b446687ae86","project":"openstack/octavia","branch":"master","topic":"fix-keepalived-check","hashtags":[],"change_id":"Ib089affa229531cd72f6853105d74b446687ae86","subject":"Use \u0027bash\u0027 in the keepalived check script","status":"MERGED","created":"2020-12-10 02:24:36.000000000","updated":"2020-12-17 17:27:30.000000000","submitted":"2020-12-17 17:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766381-fix-keepalived-check","meta_rev_id":"80b9bec2680829353b0e52e9f6dffeaf6494e852","_number":766381,"virtual_id_number":766381,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 17:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-17 17:27:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-12-17 17:27:30.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-17 17:27:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-12-17 12:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-17 17:27:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 02:43:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-12-10 09:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 22:40:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-12-17 12:51:29.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"66f318eb2468222670693ba18301f23bf1bc6633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-12-10 02:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea010102d78c1c3cdf8088702f29dab3e49b39a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-10 02:43:26.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\nLGTM\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"1f1509774a2645b488628d52286e770a013ce3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 09:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89031aceb81f4085803d2e782a8e3dcc : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb456479ea084839ab7a695cfd591d37 : FAILURE in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439e2e01f67f40a2a79009985b5e8ddd : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031f5103ab824ae7b02808ed88f52c71 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb18f4b619a44962aab6577e1e9550af : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d28da051fee64c01bdc3c5aeca4484b1 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ac51937c954486ba8410657c7194c9 : SUCCESS in 9m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/87ef5aa635bb4065a7781771a76805dc : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fda3996e8c9a4755a439d4e8c0535dd0 : SUCCESS in 6m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d700e96883e743c4b2730d23e5ce40dc : SUCCESS in 4m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96b2412f2b4f45c8b61516be541a8f85 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53a3b15c8c2f428f9ee41311594d27e5 : ERROR Failed to update project None in 1h 43m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d26dc7fbe8ee45c0be7108f55febb9d0 : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6991e3c93e144b6b869bbb34bb8d73a6 : ERROR Failed to update project None in 39m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a30808fd97bf45b28e73093e740c5af8 : ERROR Failed to update project None in 1h 21m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aee5b06253684d4ab234e5df2a12bc0e : ERROR Failed to update project None in 1h 02m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/425a2d7dd1854770804afd0ba3cd7e01 : ERROR Failed to update project None in 53m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b226d2f8d0be4189862909a1b39736a8 : ERROR Failed to update project None in 1h 14m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5881c057d402432d9c9946ea167988cb : FAILURE in 2h 21m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/571300b5cc744d1c9fcdce009ab4d0df : ERROR Failed to update project None in 1h 43m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a978b816d3646d88d627d53117f7786 : ERROR Failed to update project None in 1h 02m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/840c2011250a4327a7f1cbdd20d1cef6 : ERROR Failed to update project None in 1h 19m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6402e914fbf4442188d6c07c46097af0 : ERROR Failed to update project None in 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66b165a7da6c08736103198635c66b9e688779b5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-15 22:40:46.000000000","message":"Patch Set 1: Code-Review+1\n\nJust need gate fix to merge","accounts_in_message":[],"_revision_number":1},{"id":"955167ae8a49474988666a67d32c5ed9c1d83647","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-17 12:51:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd1ceb202b55dc5a01472aa8d93fbe91068de1a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-17 12:52:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6c584e09b5da6bdef0549e4d715907fb926784ad","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-17 12:54:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4a782d4ec26a0a16738d88e0d7ba64c56163ba20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:39:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e35bd944834145239c7a354035bd5243 : SUCCESS in 12m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6233a7936ebe4aec8baadc7760e23520 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b5125f72bd43d2afe085d7571e43c1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63989e79903c4795894a7fac4b23f4b5 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10a529f7fa0f4ddc9b693768e7939689 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3562841f0ddd450ebbd5d18f235f5cb9 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a200378a90d47d68ed5b2cb9e1c59f5 : SUCCESS in 16m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/97a477ac4d204a12892c1783e085c071 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a7ffc99cea834ef994ef19139a0c0e8c : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ba6dce79225747449473a560e80c868c : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/515879c263c04322a6d89d6ae46cf27b : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ebfa904e3b44502ae1c8d684f7acda5 : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afc7f3f7d421421680a13c1957278ddc : SUCCESS in 1h 56m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40f7517384a747cfbb8ec925ec822f87 : SUCCESS in 46m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/29d05c081d6840c482a9d1bd4ec8e51c : SUCCESS in 1h 23m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58d311200d3a4f9cb088c0ac5bbdba73 : SUCCESS in 41m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/095e0c4132ff4295bd052fcd85f56c9c : SUCCESS in 1h 01m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9d0fb4dfbab49068f34963457f3218a : SUCCESS in 2h 13m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f212530363d48feb2d2f581be16974b : FAILURE in 2h 28m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/51e39ac4f34b4250b65b9504321c2a61 : SUCCESS in 1h 45m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a2c9aae0e3614367a1324f6c6884976e : SUCCESS in 2h 13m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4c716dc84e2a4700937446e2938e9d69 : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b5a1d4d288344eb2bf89739022d767ff : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78f9e578f6cff0a6662c6a611f02d89eee4756b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:39:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f1340180250d3943c8d65c33f22341506154bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:25:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bac8d0dd15584b18a52817250bfe64d8 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbc386ca5f24a33ac2d4d86a77b2b7e : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e4d4ee3dbf246ffac2d5da9e838254c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c31311fbc6414d9c181368541fee14 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02bd164f3ea4ff3a4a416cc628aea1c : SUCCESS in 10m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/790ce685386340ee99060cfd7abd4115 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5d789fd47744ec7b06ec33e8d692129 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc4d0f10153f4c299c1dd5f8eb51ed95 : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb910063588443409de59831309c4aa5 : SUCCESS in 1h 43m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdd155731b3c4288805c850c909c93c1 : SUCCESS in 1h 24m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7577e07f4fee44afb96dd85ccd504b78 : SUCCESS in 47m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cc44363146b4149b0848dbab311d837 : SUCCESS in 57m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ada83c648cc4f7fadc7f0923a2ffcf8 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ecf8249586b1cd4d504310d15845c8b28fbc20f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:25:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6e373bcddc54eede7bc795c611b9478b4974850f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 17:27:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5e3bdb28a449c780b7dd45ff02ae11 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a86e454e7be22b36039a4a4bac6a886f523e9d3","revisions":{"6e0376d93ad34c9efcd25ff888fc0a85674aba4a":{"kind":"REWORK","_number":1,"created":"2020-12-10 02:24:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/81/766381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/766381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/766381/1"}}},"commit":{"parents":[{"commit":"773e41f2d821b121faf91d9780d52d187276047c","subject":"Merge \"Remove re-import of octavia-lib constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/773e41f2d821b121faf91d9780d52d187276047c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-12-10 02:18:10.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-12-10 02:18:10.000000000","tz":780},"subject":"Use \u0027bash\u0027 in the keepalived check script","message":"Use \u0027bash\u0027 in the keepalived check script\n\nUse bash instead of sh to avoid the error \"shopt: not found\"\n\nChange-Id: Ib089affa229531cd72f6853105d74b446687ae86\nStory: 2008437\nTask: 41399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0376d93ad34c9efcd25ff888fc0a85674aba4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0376d93ad34c9efcd25ff888fc0a85674aba4a"}]},"branch":"refs/heads/master"},"1a86e454e7be22b36039a4a4bac6a886f523e9d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-17 12:52:01.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/81/766381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/766381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/766381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/766381/2"}}},"commit":{"parents":[{"commit":"dc876eed9db5ed37da9598673f8e2ca2cbb0ceff","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dc876eed9db5ed37da9598673f8e2ca2cbb0ceff"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-12-10 02:18:10.000000000","tz":780},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-17 12:52:01.000000000","tz":0},"subject":"Use \u0027bash\u0027 in the keepalived check script","message":"Use \u0027bash\u0027 in the keepalived check script\n\nUse bash instead of sh to avoid the error \"shopt: not found\"\n\nChange-Id: Ib089affa229531cd72f6853105d74b446687ae86\nStory: 2008437\nTask: 41399\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a86e454e7be22b36039a4a4bac6a886f523e9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a86e454e7be22b36039a4a4bac6a886f523e9d3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
