)]}'
{"id":"openstack%2Foctavia~768245","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ib7fa32327b340eb3bf398dea34875605a23c9a62","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ib7fa32327b340eb3bf398dea34875605a23c9a62","subject":"Fix PK inspection and abstract raw SQL away to fix migrations on PostgreSQL.","status":"ABANDONED","created":"2020-12-22 17:55:23.000000000","updated":"2020-12-23 10:03:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf295296f684315fc310c67e309a31959f6f8026","_number":768245,"virtual_id_number":768245,"owner":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-23 09:55:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 19:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19bc95e68450383590276396b8385453fc6a2991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"date":"2020-12-22 17:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c26119fed80db13f91ba5362bba828bc9cd671c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 19:00:10.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05c940b00d74f42a0fd186c2cd94a5b : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43fe1158906645cf8d9d8143274363cd : FAILURE in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b06188c59049929588f310e1700bf3 : FAILURE in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808a8e282b16449db9a90f23c9309fa7 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c02b33e26f2494ea362b5c7999b6766 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ba2b87bf2d49b0bcffff4ad1e888d8 : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27240816551348cd979719b12d9d2c1c : SUCCESS in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ff46284691ca4e29b351495dfdafda9a : SUCCESS in 16m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a85a85434a1a4c30a12e97b686f1dc57 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fc6ebe69d2f49178df19a23a857fd71 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8efa45b23f334b518d4a8432194068b0 : FAILURE in 22m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4121d2559faf46da8ab4a713dec0c1f6 : FAILURE in 30m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17931a5ca6694e4cafde63561f2900d2 : FAILURE in 23m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/650a60835f1541349a82ef0182227846 : FAILURE in 19m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d98371e5573940949fed4f2987ff3f4f : FAILURE in 19m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf6a47bb37f44d08a454f9f5317795a8 : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ea228a896f443a68bd36abbe5fd4f1e : FAILURE in 22m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4923ae5b79b45cdb16b95f9029e4088 : FAILURE in 22m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/29fc170f4e2c4034a9eae5528a5e4f57 : FAILURE in 25m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f78b15083916482fac1d871231103990 : FAILURE in 19m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ac99577dfb364ff69ae0840ff6d9fce5 : FAILURE in 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f20dec8e4936434e9027ed9293a164a9 : FAILURE in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01d7e9a150e08b997b0eaa2a19d1c7bde645bba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"date":"2020-12-23 08:46:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18c955978971a49c0f5795368865e94c514d2065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"date":"2020-12-23 08:50:05.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0264013460f38a7db1faaa3c4d10491ec2d45d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:55:38.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66e2214463664da88ff1d921f3480607 : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0135eea7b27f461f8075ef48085cee04 : FAILURE in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7fd07e3e3424eb98ed9798853a250af : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5225a944897405fa903fe415370f42d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f4ec960a590451c807805f7f340db66 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50cf736371c846f6a3fd233ccb29d955 : SUCCESS in 12m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9ab28e94e3444d9e97acce78dd6bdaf1 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/967f107be3604d6191ada906f5e80876 : SUCCESS in 8m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7733bbb9e2b476aa1f3a500b0750b89 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a0f0f2f867d464a864c4ce8112256f4 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ccebcffd536f455bb47ad08cac1ef4b8 : FAILURE in 51m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58dc74e015764bb189c759fb4cd07f28 : FAILURE in 19m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47cd222c836542dcbb9925c95ab4b337 : FAILURE in 23m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5379b72d813445ababca76df74348fdc : FAILURE in 21m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e448eb9caf054cb18a9e6c890adcbd53 : FAILURE in 23m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee87737512df438994983e7e8909d2d6 : SUCCESS in 1h 02m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c657ee19547d4309b4894db7f1cd76ee : FAILURE in 24m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0163da703434ce8921197d9b274f0cd : FAILURE in 23m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a580fb3c4a0d4c94b10618d7bbf6ef59 : FAILURE in 23m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b254c3d59394423a87aa979514c1f9d : FAILURE in 22m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d81249aa322d49099d7eedabc83e9958 : FAILURE in 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4b09637a2dd94ffeac28a7fdf8ada07c : FAILURE in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf295296f684315fc310c67e309a31959f6f8026","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"date":"2020-12-23 10:03:25.000000000","message":"Abandoned\n\nRetargetted these changes to master.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"58455bfb1a216f588f67db9c24b58e37598c5a8a","revisions":{"e6f5826f3cab4bb9f68d031d94977175f19e729b":{"kind":"REWORK","_number":1,"created":"2020-12-22 17:55:23.000000000","uploader":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"ref":"refs/changes/45/768245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/768245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/768245/1"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-22 17:53:09.000000000","tz":60},"committer":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-22 17:53:09.000000000","tz":60},"subject":"Abstract raw SQL away to fix migrations on PostgreSQL.","message":"Abstract raw SQL away to fix migrations on PostgreSQL.\n\nAs a side effect, this might also enable applying these migrations\non other non-MySQL database engines.\n\nChange-Id: Ib7fa32327b340eb3bf398dea34875605a23c9a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6f5826f3cab4bb9f68d031d94977175f19e729b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6f5826f3cab4bb9f68d031d94977175f19e729b"}]},"branch":"refs/heads/stable/victoria"},"1843fbc47c8a964ebee990c857355a3b0a723bcc":{"kind":"REWORK","_number":2,"created":"2020-12-23 08:46:32.000000000","uploader":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"ref":"refs/changes/45/768245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/768245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/768245/2"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-23 08:30:11.000000000","tz":60},"committer":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-23 08:34:53.000000000","tz":60},"subject":"Fix PK inspection and abstract raw SQL away to fix migrations on PostgreSQL.","message":"Fix PK inspection and abstract raw SQL away to fix migrations on PostgreSQL.\n\nPrior to this change, the inspector created a separate connection\nto the database, but since the schema migration\u0027s single transaction\nhasn\u0027t finished yet (and therefore, hasn\u0027t committed it\u0027s changes),\nthis caused schema revision 392fb85b4419 to fail. Re-using the same\nconnection for inspection provides it with the same view of the timeline,\nprior to committing those changes, resulting in proper migration on pgSQL.\n\nAs a side effect, this might also enable applying these migrations\non other non-MySQL database engines.\n\nChange-Id: Ib7fa32327b340eb3bf398dea34875605a23c9a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1843fbc47c8a964ebee990c857355a3b0a723bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1843fbc47c8a964ebee990c857355a3b0a723bcc"}]},"branch":"refs/heads/stable/victoria"},"58455bfb1a216f588f67db9c24b58e37598c5a8a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-23 08:50:05.000000000","uploader":{"_account_id":32871,"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","username":"ksdziekonski"},"ref":"refs/changes/45/768245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/45/768245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/45/768245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/45/768245/3"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-23 08:30:11.000000000","tz":60},"committer":{"name":"Ksawery Dziekoński","email":"ksdziekonski@gmail.com","date":"2020-12-23 08:47:15.000000000","tz":60},"subject":"Fix PK inspection and abstract raw SQL away to fix migrations on PostgreSQL.","message":"Fix PK inspection and abstract raw SQL away to fix migrations on PostgreSQL.\n\nPrior to this change, the inspector created a separate connection\nto the database, but since the schema migration\u0027s single transaction\nhasn\u0027t finished yet (and therefore, hasn\u0027t committed it\u0027s changes),\nthis caused schema revision 392fb85b4419 to fail. Re-using the same\nconnection for inspection provides it with the same view of the\ntimeline, prior to committing those changes, resulting in proper\nmigration on pgSQL.\n\nAs a side effect, this might also enable applying these migrations\non other non-MySQL database engines.\n\nChange-Id: Ib7fa32327b340eb3bf398dea34875605a23c9a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58455bfb1a216f588f67db9c24b58e37598c5a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58455bfb1a216f588f67db9c24b58e37598c5a8a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
