)]}'
{"id":"openstack%2Foctavia~768464","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I5edd05f549d4317b2c6186c35c6fc64cb54facac","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I5edd05f549d4317b2c6186c35c6fc64cb54facac","subject":"Fix lower-constraints","status":"MERGED","created":"2020-12-24 11:02:31.000000000","updated":"2021-01-08 00:22:59.000000000","submitted":"2021-01-08 00:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768464","meta_rev_id":"7504ef0454be6398f1d7c6fa1819f544b20f23fd","_number":768464,"virtual_id_number":768464,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 00:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2021-01-04 17:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-05 08:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2021-01-07 22:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-01-05 08:31:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-01-07 22:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 13:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-30 08:06:04.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-12-30 08:06:04.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-12-30 08:06:04.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-04 17:23:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-05 08:31:42.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"6c9b4e55ee34eef49b7728527101854fd49c81c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-24 11:02:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c660b663239187db945d4e830f52510b0a4ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 13:01:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/adbf2b6975834a6c85325c484da40fb2 : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3db93494f842d4a9055af1afa19644 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337ac5519e8d41b59e55bd0a7d5866a0 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39c1f9c8eba41fab7c490da3c04349e : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee808a65fb9f4b2b8941eff6ccc491c5 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c238ae866314ad8b812a7ba47cb1ded : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb0db785f9643b2ba88e74f4ba27a0f : SUCCESS in 10m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5876d30df0494a128acc9c619a6fc227 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/216a49612f294e4ab8edda577bec5ab6 : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9848d52b33e440f8a3bafd2529bffd65 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a6d61f073d34afc9adb98a29ff57600 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/feccd7860e4046e1960bd1564488a195 : SUCCESS in 1h 17m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/062d5fff40a848ca85400f3a36cb7103 : SUCCESS in 1h 32m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b3e41935784a4eb6a31088da6397e953 : SUCCESS in 46m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bee29cccb27841a292e5b4af35c3ee49 : SUCCESS in 38m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d58c8413598a487491d59f7a38ba7b81 : SUCCESS in 37m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/45cf6866eb034877b988572d82a1c6b9 : SUCCESS in 55m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5525fc0940244f94b5aa14a234d68ea3 : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7c6b4d6b9474b76ae85ea6f98a553d5 : FAILURE in 1h 56m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d3c4a7817804bedb7aea117492f59cf : SUCCESS in 1h 29m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a39bcddeaff498fbb069fc132b826f9 : SUCCESS in 1h 32m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3afa3915d8924198a0445055076b1a35 : SUCCESS in 1h 35m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4f70012aa8fb4f60830c1ac95b85cba4 : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ea2a2d23cb43d926b3ad365892cfac7174be37d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-24 14:06:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0a76e72add644da481ff7f194f4cfa62555240d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 16:11:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/44b664e17bda46918e649368688a7cd0 : FAILURE in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e203a267b47494a900ff970d1182a68 : SUCCESS in 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67697e13e9ca469bbebc0a0d72473e42 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8adbd99d3d524b0bafe42548e0c6955f : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f50a67599e49f39b60337dde16ce9e : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2be85399c7041ac9a000f393879060a : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d130ef4ceac84606bd28235b9cf22775 : SUCCESS in 11m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5997ddce0ffc441aae53701e8f8f8624 : SUCCESS in 4m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ec8a53ad38b645839e0a14536cd45562 : SUCCESS in 5m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e50cbe57bf640ff9842a529c09de8b1 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5b4e0a473f248458b2f1dbfff7bd33a : SUCCESS in 5m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d6d1ff1427745448669e7e91cac1a5c : SUCCESS in 1h 28m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7fe9bc42cf74486a02dd08b6828be66 : SUCCESS in 1h 31m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c7d5721274ed456bae570d1420da5ef0 : SUCCESS in 49m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2eff20e35ee43a0b7697c683c3eff7b : SUCCESS in 52m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46ee2480ec00406189d8fcde5f37ede4 : SUCCESS in 38m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6565fbee0498466fb00dca5b24748e7e : SUCCESS in 54m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5db7bf0c44fa4ac3be3a97d59c7d8d1f : SUCCESS in 1h 38m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b41c634db964707852e4c47a7f87eed : FAILURE in 2h 03m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50b6c8b0bbd641a89afdbe28f69c9749 : SUCCESS in 1h 30m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a5c7e584e73432ebbc1dcade37d0cff : SUCCESS in 1h 48m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eae64f4066dd41bb97e6d9b9b818e6b3 : SUCCESS in 1h 29m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3895161e2eb24eff8f9086b0f97fae02 : FAILURE in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a04e5f0d1c0735a540e39c61bddab57201a698b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-28 07:29:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d21bd88f97359575fa85b6588d94dc8112ff8b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 09:39:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2012ba22fb5c4a65be88086ec01e80a6 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b0a59984b524c7cbdcf8f60775c04cb : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc19771d73540efa5e5d58d37be4f3c : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd2e0178059f46a9a272d7c13628b495 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c26fa5fa5896491dbd6093e1fbeb5422 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37c0fd5df80749f1817812b65984fcc0 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44611bc03ea74d3d9aa52c9f2211666f : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e3a54ec42f54568877253308bae0a5c : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3398704afa624df58faea6d6845e6f7b : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24e616fb264545ec8a0a6dc7ebb2a636 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c2c15c5f3b043769efd843925da8004 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d0698f05c84466ea5af37c171ae385a : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6935042584bd457cba0ffadeca66aa31 : SUCCESS in 1h 19m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c6b879ef654fd0a27f814b275c0ab6 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d0f1969c3124015881aece4eb464bae : SUCCESS in 47m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/73ea97b0fac74b108c02838b632e04df : SUCCESS in 44m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8c974263858461bb098f35f239b2a7f : SUCCESS in 44m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e1c9a0c324c4b4f83a01187606d940b : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e67a363b02a4435fa81a219eda9e7dc3 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9cff4bc65d4455ba876d88c53bdfcb9 : FAILURE in 2h 07m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e0d78948fce74d8c8e59080715fab820 : SUCCESS in 1h 38m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c47eb44fe55a460bb9c0a4dbe915cce1 : SUCCESS in 2h 02m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/90e8d5ec70264d24bcffec4e42ad7172 : SUCCESS in 2h 08m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cf0bd0bc164e4d588b485121e0e138b4 : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e03270b6515047b40dde39b529a44b33036a49e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-04 17:23:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ef8086503348a9203ae7d054b6815a574648f256","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-05 08:22:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cd03d980c8b2b1be2d180cf37045a383da64c1c4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-05 08:31:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"050f424439f1a303a234d7dad33e6bac09dd9c46","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-07 22:32:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI am not super comfortable with bumping requirements on stable branches, but given the python environment issues, I think this is a required, yet minimal, change.","accounts_in_message":[],"_revision_number":3},{"id":"afeb7434193cdbbe860e6da1f09644b730a267ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 22:32:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d8980df4c195e1eee325a967b53012dce203e60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 00:21:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b5402e7a9d394340bf6dc85bdb8bf63d : SUCCESS in 2m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f734bca769e4c1abb4a03e996dc7854 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf200fc5964243fda942cd54bc702dfa : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/037abc7c96b4450eabc3a1b85d687deb : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08943fe7ec9f44e1bc9e1e22b4da403f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01013f50d2664ba1ac45dd727031d21c : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/242ff6a2ef034a4d9910d1fa4dd40791 : SUCCESS in 5m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cb95cb92f38d4b999a3bd2a136b2f721 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3631b107179424781ef1110eb4be5ce : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57a43ef2c91640b98110899354b6105a : SUCCESS in 1h 12m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/095ea536ffa64bf8a7333273573fa398 : SUCCESS in 1h 44m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d14d6642d5164f55a8a5382b0b8eac8c : SUCCESS in 35m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/357398a180a149269b5a10da9831ec9c : SUCCESS in 1h 29m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79f7f344ec4f4237811d01c5b7e68ee9 : SUCCESS in 43m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/968754dd23e44405a7433b719fee27a3 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6eaea19d6507b4c04d7197d466e21ef80cc90bb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 00:21:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7504ef0454be6398f1d7c6fa1819f544b20f23fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 00:22:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d2a48c200e4379a2009885c0e58455 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4153906d3c7d45949ccdf745344a3c10 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb","revisions":{"90a634b521a4fcfe082532b90ef7f0d3e5cdacf4":{"kind":"REWORK","_number":1,"created":"2020-12-24 11:02:31.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/64/768464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/768464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/768464/1"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-24 11:01:10.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-24 11:01:10.000000000","tz":240},"subject":"Fix lower-constraints","message":"Fix lower-constraints\n\nBump hacking and pylint requirements to unlock gates.\n\nChange-Id: I5edd05f549d4317b2c6186c35c6fc64cb54facac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90a634b521a4fcfe082532b90ef7f0d3e5cdacf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90a634b521a4fcfe082532b90ef7f0d3e5cdacf4"}]},"branch":"refs/heads/stable/victoria"},"a44342f39294863e1b284fd32470273f7f918ad0":{"kind":"REWORK","_number":2,"created":"2020-12-24 14:06:51.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/64/768464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/768464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/768464/2"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-24 11:01:10.000000000","tz":240},"committer":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-24 13:02:32.000000000","tz":240},"subject":"Fix lower-constraints","message":"Fix lower-constraints\n\nBump several requirements to unlock gates.\n\nChange-Id: I5edd05f549d4317b2c6186c35c6fc64cb54facac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a44342f39294863e1b284fd32470273f7f918ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a44342f39294863e1b284fd32470273f7f918ad0"}]},"branch":"refs/heads/stable/victoria"},"e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb":{"kind":"REWORK","_number":3,"created":"2020-12-28 07:29:09.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/64/768464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/768464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/768464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/768464/3"}}},"commit":{"parents":[{"commit":"9471d05976495b7f35115bc1f7727df781e1cbeb","subject":"Merge \"Map cloud-guest-utils to cloud-utils-growpart for Red Hat distros.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9471d05976495b7f35115bc1f7727df781e1cbeb"}]}],"author":{"name":"Ann Kamyshnikova","email":"akamyshnikova@mirantis.com","date":"2020-12-24 11:01:10.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-12-28 07:29:02.000000000","tz":0},"subject":"Fix lower-constraints","message":"Fix lower-constraints\n\nBump several requirements to unlock gates.\n\nChange-Id: I5edd05f549d4317b2c6186c35c6fc64cb54facac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
