)]}'
{"id":"openstack%2Foctavia~769012","triplet_id":"openstack%2Foctavia~master~I902fb01ee0a62c5399e495aacc961fd254f07b4d","project":"openstack/octavia","branch":"master","topic":"sgrule-protocol-none","attention_set":{"29870":{"account":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"last_update":"2021-10-26 09:38:47.000000000","reason":"Gregory Thiemonge replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I902fb01ee0a62c5399e495aacc961fd254f07b4d","subject":"Fix vip security group update","status":"NEW","created":"2021-01-03 08:16:40.000000000","updated":"2021-10-26 09:38:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"d697ae26f7ca81e5caa98011e9d1d445267cd5aa","_number":769012,"virtual_id_number":769012,"owner":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-06 03:13:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2021-10-26 09:38:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 10:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 17:17:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-10-26 09:38:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"22c25e8035a769c054040b37e31f6a0d63e3e835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2021-01-03 08:16:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3479de5f29889d59ef9279dfd799e8a2f5121fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 10:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7d6ad696b440f8abceb751f55f8cdc : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a52e2c5baf843a0b9f3ddc3dbf24ea3 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76193efdc2484b5fb10138c2b64fd99e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84813cdab855415eaeeb113d9535b4a2 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/039d85e70aa748258478de627c6e9370 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5451df27c3fa41ec9a119dbab9bc46d6 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6473c4d30254149bff8393bc2c2f6f3 : SUCCESS in 11m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7315f86fac9144869267c66c8afcb9b4 : SUCCESS in 5m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b0d694a8db2f449eac826c41eff2b6ef : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2832f343e721466d8f9f34dbceb4c3af : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5aeafbf13764b22b3c949fc1c4e09ac : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f5f137b393d490e8e4bfd127e74ef97 : SUCCESS in 1h 18m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f3a39f5121d4ea496dc9a2333961f0a : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3cd7327f8cd74a3eaf389f03825f9b5a : SUCCESS in 39m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/010e696016b94fc8809dca2bd3e4c90f : SUCCESS in 40m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f424a959f3b469f80625034419746e4 : SUCCESS in 37m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/09db74fec61f4854aa3f88009731bf4d : SUCCESS in 54m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9086fef769b348e387a51cc1589cb934 : FAILURE in 1h 39m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad62ec1b1ad146a697875632bd34803c : FAILURE in 2h 18m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10ed2169127f421198b532120e98eef1 : SUCCESS in 1h 51m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d450f502ad1f4406ad09b83788bf3aae : SUCCESS in 1h 57m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2994c7db76e142de80c7bb5036c6018a : SUCCESS in 1h 44m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5bbd3bea65c84a1b98cb416275627cd9 : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"561d0b0dbfc26a8d496b607c6bb6cb4830656b61","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-04 17:17:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nShould open a bug/story for this.  Also, I think it would be easy enough to add a unit test (or add to an existing one) to catch this issue.","accounts_in_message":[],"_revision_number":1},{"id":"777b8aeace8d6305548b3341366f95f29e1087f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2021-01-05 11:22:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fee20f113f4d23aa4d5e951feb9e0aecde5f6644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 13:32:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fcf00d308004807b58cfabd0e44e7b9 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2784e98d0df74b8bb60f7dad239c0ecf : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630f8cab475c4416ad0b8d7bb1f56425 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e69a76d78e29488b977ac363e5190e7a : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4ac0ff7d1e44f6baadfadb77e6f862a : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbe6eba7b25a44938149405da8c9940f : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b83617e61e94d718f86c373d834a867 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7acd5bdf4f484f299eca947d43279385 : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31b11727ca8442eda2382d41447295ed : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b00447764e0470fb4e652e6c5ff8969 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/554c29e607dd4a0f8338469d589ad59e : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6b1f26797e742019b7d8f238fdb762a : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6be5c513fd34bb6a3071d813cdabea8 : FAILURE in 8m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2a446c5b272445a3acdd580082283831 : SUCCESS in 40m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/79d11ee2e966477da95518e63296007a : FAILURE in 9m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7b0074ea14a42a7b64ec56b5cd19f24 : SUCCESS in 41m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/722b92ce4f4b4611a02e4254437ea0b5 : FAILURE in 9m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60e2a3f20f8f4a9fb126f688337a3458 : SUCCESS in 1h 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/261b6aebd38143cf9bcfa21adff053e9 : FAILURE in 11m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82371f9470aa4327824dbcd4f69aaeeb : SUCCESS in 2h 07m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ab12fca940714897b9c1d18475ce7037 : SUCCESS in 2h 04m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c80f9bfd9cc24d39b9f0fa8d3f45edec : SUCCESS in 1h 56m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dce2473a00fa4f729a53c03b32915ab1 : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4d7444d248e492168d6045ede1cf3c524bcbe11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2021-01-06 01:11:48.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5e01fc97f66b68350c032daba58e7e87b013d49e","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2021-01-06 01:35:42.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Should open a bug/story for this.  Also, I think it would be easy enough to add a unit test (or add to an existing one) to catch this issue.\nBug/story added. And the case we try to exclude is that the sg rule protocol value is None，so we change a more precise writing.","accounts_in_message":[],"_revision_number":3},{"id":"370fff5484d7d4eed4d8477a13a4403a7c296994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 03:13:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a6cd3c2970e4b729f4367d856c2cbdf : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4237665d6d8d444094b59f111e1f37e3 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a08693e95a4c119a6d8182b9386be2 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69814ba72c2145ed817c8b70a1d312f6 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92e3514566034ea983f65628b6b12316 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d254929b6ab4c5a95e72259e7f349f5 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ee1c3b52024a719a5ce8dedb5d62db : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/989854efb4434b4d9906aa8f8b27c663 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a260ff599875433ebe986e0e1cdaf3a5 : SUCCESS in 9m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/671a70e7caa14d3fbf22e6df5ec893fd : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a08d7fea3e9e4501a523a67330dcd7a3 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d312e9cab2ef40c5be0fb8b7db51f256 : SUCCESS in 1h 05m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce76ec690a4a401ba03dd811903221c0 : SUCCESS in 1h 46m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72d8e00990b3434785b73d92e133e9fe : FAILURE in 9m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/326531202b6d4c85b4b1332a58ad3823 : FAILURE in 8m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dae9a6d0a3c24f07a3335182fd1fce2d : FAILURE in 8m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19e430fa43834d3586f7588af65b3314 : FAILURE in 9m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2dda31d05604eb597df9ee9f93e8225 : SUCCESS in 1h 56m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c06e07df8867427daf8f8d9e20c29f0e : FAILURE in 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c0e4d63e0cd4d2ca114d60fb873fd0d : SUCCESS in 1h 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b64a77ff409945aca411c575ee417ba0 : FAILURE in 2h 00m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c0f8fead4d1a43a9908201e02965d0b2 : SUCCESS in 1h 39m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fd5f83554a4d43a29e1f75b4941887f2 : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aed7de54401b457455dc257237aeff54e0b9b2e7","author":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"date":"2021-01-06 04:40:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"92379ca11ecfa7464c1fa3b0b4cfb14af2a34e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 07:03:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10be84a144da47d6a667cd5f2bbf323c : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a89d613a1c6248e586158694fb833718 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3788df52a0d048d5aec47d457209e64d : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/900a1bf743334fdca641675cc01a0a90 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ac5b23a75de4a60bb70fbac713c8d40 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4ac9b3d2224aeaaa9fc4de3bd5ed69 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9866a496034fdbb59bf2329472d646 : SUCCESS in 14m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b80991905844ca3bfa82f5069e7b7c8 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/10137d83da7b4d99b92e3b25271117f5 : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/975f1bf2bc24414e989121c0316dea43 : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/131a81b8089243c3b5aeb1d53bafb503 : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4025e39e1d843cbbee94b2f5d9b1130 : SUCCESS in 1h 05m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55ce07a1177d41639412c36f284f22b8 : SUCCESS in 2h 20m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ca94f56287645709e24bbd92125c914 : SUCCESS in 1h 01m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ce9751c350384e769f0902a6b4a8201b : SUCCESS in 1h 00m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9baf257646a5480aa75da4266ab0ed2f : FAILURE in 1h 42m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1350e2a5881f446695fd7d22b40f7d5b : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e509b5d930fd4ab08f4c62161f8ee279 : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6973515d763c400298894113d3fd95db : FAILURE in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0316b0dd771742068ec4ffacd2df2fc4 : SUCCESS in 1h 47m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4339cff7cea4a17bfe375ad41885c8a : SUCCESS in 1h 37m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d2f84962aa3e43c5b7f6114fcf36e8a4 : SUCCESS in 1h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a98523a620d340ee935c34bcc0e7a2d9 : FAILURE in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d697ae26f7ca81e5caa98011e9d1d445267cd5aa","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-26 09:38:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"154e826fa8ff4fc530f34f778873dac4001099ea","revisions":{"7dc658c7a78ee878469a526f88341b7c7658d717":{"kind":"REWORK","_number":1,"created":"2021-01-03 08:16:40.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/12/769012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/769012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/769012/1"}}},"commit":{"parents":[{"commit":"1a86e454e7be22b36039a4a4bac6a886f523e9d3","subject":"Use \u0027bash\u0027 in the keepalived check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a86e454e7be22b36039a4a4bac6a886f523e9d3"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-03 07:59:55.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-03 07:59:55.000000000","tz":480},"subject":"Fix vip security group update","message":"Fix vip security group update\n\nUpdating lb will fail when the lb vip security group rule\u0027s\ningress protocol is None. In particular situation, it does happen.\nWe assure the updating process normal regardless of the rules.\n\nChange-Id: I902fb01ee0a62c5399e495aacc961fd254f07b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dc658c7a78ee878469a526f88341b7c7658d717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dc658c7a78ee878469a526f88341b7c7658d717"}]},"branch":"refs/heads/master"},"09a1e67b00d68d07f1b7fa4a9d49d6ddbadd5f9e":{"kind":"REWORK","_number":2,"created":"2021-01-05 11:22:54.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/12/769012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/769012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/769012/2"}}},"commit":{"parents":[{"commit":"4704e458a8f0c5886977aa4c87a6c805a518c43a","subject":"Merge \"Fix periodic image builder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4704e458a8f0c5886977aa4c87a6c805a518c43a"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-03 07:59:55.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-05 11:22:35.000000000","tz":480},"subject":"Fix vip security group update","message":"Fix vip security group update\n\nUpdating lb will fail when the lb vip security group rule\u0027s\ningress protocol is None. In particular situation, it does happen.\nWe assure the updating process normal regardless of the rules.\n\nChange-Id: I902fb01ee0a62c5399e495aacc961fd254f07b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09a1e67b00d68d07f1b7fa4a9d49d6ddbadd5f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09a1e67b00d68d07f1b7fa4a9d49d6ddbadd5f9e"}]},"branch":"refs/heads/master"},"154e826fa8ff4fc530f34f778873dac4001099ea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-06 01:11:48.000000000","uploader":{"_account_id":29870,"name":"sunxifa","email":"sunxifa@inspur.com","username":"sunxifa"},"ref":"refs/changes/12/769012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/769012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/769012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/769012/3"}}},"commit":{"parents":[{"commit":"4704e458a8f0c5886977aa4c87a6c805a518c43a","subject":"Merge \"Fix periodic image builder jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4704e458a8f0c5886977aa4c87a6c805a518c43a"}]}],"author":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-03 07:59:55.000000000","tz":480},"committer":{"name":"sunxifa","email":"sunxifa@inspur.com","date":"2021-01-06 01:09:41.000000000","tz":480},"subject":"Fix vip security group update","message":"Fix vip security group update\n\nUpdating lb will fail when the lb vip security group rule\u0027s\ningress protocol is None. In particular situation, it does happen.\nWe assure the updating process normal regardless of the rules.\n\nChange-Id: I902fb01ee0a62c5399e495aacc961fd254f07b4d\nStory: 2008484\nTask: 41532\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/154e826fa8ff4fc530f34f778873dac4001099ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/154e826fa8ff4fc530f34f778873dac4001099ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
