)]}'
{"id":"openstack%2Foctavia~770176","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I30e23ca13d033d77c8ebdabbfdc7b54556a9466b","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I30e23ca13d033d77c8ebdabbfdc7b54556a9466b","subject":"Fix operating status for empty UDP pools","status":"MERGED","created":"2021-01-11 18:50:55.000000000","updated":"2021-01-15 00:53:59.000000000","submitted":"2021-01-15 00:52:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770176","cherry_pick_of_change":763530,"cherry_pick_of_patch_set":2,"meta_rev_id":"be4441ba9d47d99037fdcc16621d16b2ad7b2b75","_number":770176,"virtual_id_number":770176,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 00:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-14 08:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-14 08:25:54.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-14 08:42:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 18:50:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-11 21:11:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 23:46:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-01-13 23:46:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-01-14 08:42:18.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"b5897b07473afe6d969b0eadebff01c4cfc93a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 18:50:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a939509cbb99b005c5e4c412583c62d0c8d579f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 21:11:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36b344ef21b44718b346f9d8f538e119 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11704dab7c924200aa97cb40484ada32 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc56b7bc88ed4232bf94db5ee15a34e8 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec78044b39c49738bc44301d4b67edd : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50f291705f14d079dd9089f239d3adc : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd2ef2b228b4f209db602f3e9f32767 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5ec69b80534b13b52679bb5fee8824 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8a5bc74eed8b418ca537762a377ddd51 : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9b5513e7db9d42caa6a2d519275b6916 : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c94b3933a62c4ccebe5bae26f0a5cb1a : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6684e9aa2b904b3b92fe5f1b9204a86c : SUCCESS in 8m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/206d421d19ff4d3a817feb259d15c51c : SUCCESS in 1h 17m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b912d09bded84417a5f9b6d9ee362a10 : SUCCESS in 2h 11m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25d24ce17e034b6099d7043369498f86 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b49c689a6444ba3a2601fb10209e914 : SUCCESS in 39m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edd71211fab04f01af4a8c78c80e8aed : SUCCESS in 44m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e7f2ddeb6024d71909fc5039b1797c2 : FAILURE in 30m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3152dcd774bb4f62af7ed94c7bdbcef2 : FAILURE in 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/93e05f9708694f9fa4848f77e52a9750 : FAILURE in 28m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f758632188a342baabf330fd7c9f1095 : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0797d7514c474449a2a8099025f856fa : SUCCESS in 2h 04m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5fe3a7181e6d4f0fb6f89ff5f1bb6ca9 : SUCCESS in 2h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1131682f03ab44acafa7308ba07e9bd9 : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73a514234acdb8011215433eb5f3b4801c8a03a4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-11 21:13:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d52814dd2c37792340874db46559a2776c8f8f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 23:24:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec4e2d2b36cc47628dc6d05bafdcb5ae : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2698316b69c743be95523d5fc43b61de : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81cf89fe828c48658a6047a195e6f9f5 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55cabfa7b8764d28af0770e6dd93aff4 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59da6c0bd7fb4009908aaf574d04e59b : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6f753c04ae4c93a48ed4533de6c120 : SUCCESS in 14m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e2bb2cff6bb47d58a58531a45809ed6 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06159f8d515e463197062949b65c275c : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5ea8212d873d42d986472ea694600709 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56e983def3014430a1179d468f3948f8 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36372ce642e44e64ba30b3262725149e : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53b31e6702df42259054c336f479b6f3 : SUCCESS in 1h 14m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c10a6b2add7e4d9bba5479cf78181267 : FAILURE in 1h 53m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe608445579c42dfbbf4532fba0737e5 : SUCCESS in 44m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0bb224f1e7ad4d8c995526050d539e24 : SUCCESS in 1h 08m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56ebeea7cb164414a94389b0d68fef80 : SUCCESS in 56m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/313de1c5864846cf97596e1f3989bfc7 : FAILURE in 54m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6592e724133a4858b4a46e4caa429803 : POST_FAILURE in 1h 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b858e6153cd412fb23a4a66fb2596aa : FAILURE in 2h 05m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8b9f1c72adb540ef91a5a5756de208d5 : SUCCESS in 1h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4d05dd4ff8464352b2b7ef6d85295186 : SUCCESS in 1h 47m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/228cde76b9e741e1b8d11048cfa35962 : SUCCESS in 1h 49m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/82d0d0d6235543269b65ecd73e6d6194 : FAILURE in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f61c328c8fe56a104172926c6c5309c3fd6d29e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-12 15:37:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11be53e2afff7c0852252423f7c0e47e24d29265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 18:21:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e56bc4c2014eae98591856d29593f8 : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fec98a8c5c144a6a8197b43c9bcda3f : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d9f427777345d9b9cc042686c492a2 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eecf23fab7b4d43bf138a872a778dfb : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb22ebd3e3944b3db2e31078e0cc33c9 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1269e0daf60d447abac928f383e586f0 : SUCCESS in 15m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31d7e76481f44aaa8b52db49b1c274b0 : SUCCESS in 6m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/acd34f6a09154a528bdb805bc6cc5e34 : SUCCESS in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/981a085a87c0424b97d4fb03c527d444 : SUCCESS in 8m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/820bc601d7e7497786569d2c4ea84213 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28856df837cf4b608842a042e6294a5b : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88f137bd4fc54466b8cfc7cb46b1a511 : SUCCESS in 1h 17m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1e7ff7ddc914418ab67364cafefb011 : SUCCESS in 1h 34m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3575c3c481064713af512f449078e130 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c533ad398eba41d8b8c8281e50d86be7 : SUCCESS in 42m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/479419cecc25400e8369859b0f6eeead : SUCCESS in 41m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bda72a225a564c9ca47b1a5aa7c9264f : FAILURE in 26m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8ce34cc8c3754d5db062c257270020d1 : FAILURE in 2h 20m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42ebd272248d4f55a9e73be8f876849e : FAILURE in 1h 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d6f586829d7c43678b27814e642a496c : SUCCESS in 1h 41m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a96474260a1d4a2fa0d87881410d8980 : SUCCESS in 2h 23m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f8daeef70544b768ffb3468137ce661 : SUCCESS in 1h 41m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/360e8bb247ed42b3aa8a2a35d42f797e : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98f3af807bec7b9d5f317e44b992b2ccddb117c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-13 16:53:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca32501ff23105dd0ac8966b4809350c50041893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:16:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66b67cfd2b914032a279c8d310fa1567 : SUCCESS in 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d6f1100af7d4ff0a18ecdec8b7e19a2 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2703941b6ad4aff91e4e9995264b053 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9083c2f2bdc9453a9d4c85ce409f2c45 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d33381596a94485a15fb0f067c511b8 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90691770a85548f1bd0862aeff1c9929 : TIMED_OUT in 31m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b449d67c679e4f8f9d5a8c2faf0d466d : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7c556a9e0be44d1aab5ec775561b4f58 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d344a71cb9814041ad927a392f5e100b : SUCCESS in 5m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/753d798fb5d443ff95d0394e7d230fef : SUCCESS in 9m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee07b73a5bda4f32b8356c83645fda8b : SUCCESS in 13m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b8c96973f1343b09c58a0e4bd7e75da : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0f3df2d2fc14fbb9dff0f8ccdfe2264 : SUCCESS in 1h 50m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba547114e1e84a56a9bfef999d986810 : SUCCESS in 39m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5e6fe6fa0e5449caae0fcdd3e828a8dc : SUCCESS in 43m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17465ec9ba4146c5a72984ec94c7b1de : SUCCESS in 41m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a197a7781cc463cb9f681224ebcd4e0 : SUCCESS in 1h 13m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c473199ec18c4969b547e4893beafa17 : SUCCESS in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96f0de78a1164f5da8e08b13115051d1 : FAILURE in 2h 01m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a78c7c7f3cf47279b78d8667226f961 : SUCCESS in 2h 03m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6cd4e2055ac14db09dabf3bc3f6b4fcd : SUCCESS in 2h 09m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/89bbc1539c35433e8c2815400115506b : SUCCESS in 2h 05m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/13775ec44cf9484c8894bb5d4b505600 : FAILURE in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62698313961490a54a70a57a81eff9378f8aa3e7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-13 20:55:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ca33a39fc9858c02adfcd9fa07ef9facd66d6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:41:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa73a1689b5e46dbb8fdcbf2731a9a2c : SUCCESS in 8m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c4d3b0dec044074b36c96e0b3827def : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ac148476604a53a2af69da379ef719 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d76067eac9e4bf3ac38af39f620b963 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183c42b28e9a43bb9460a4ef1b01a05a : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/599d45db5ee14e7ebb6021bb86d3b677 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2edd6f28cd6a49fda155445a747f2802 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0631a52fad314010aed4e3985636ed05 : SUCCESS in 6m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3660e466b3bb4e94acd9124ede59e370 : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a8ccd2a7d13c4c538b8d720b5b5cd2a4 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d0eb4bf73684c89941f79cacfe3b118 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/919f16af890c4a35bc80d9f59ff64ab4 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5777a6b343c34c088d6818d7d774cba6 : SUCCESS in 1h 56m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68d208e91c094a01be002cd02a3bf68b : SUCCESS in 45m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ea8119b6702c48cfae6002091179eff7 : SUCCESS in 34m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b61ae37ccc6142a8a6772ab453c73615 : SUCCESS in 40m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a22ec9d1a9d24785a875d119b915e0c9 : SUCCESS in 59m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e629e6ebecba4b1abf5a4d8f26213be0 : SUCCESS in 2h 10m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14faa97abf1149db8a837eb789eabfc6 : FAILURE in 2h 28m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d2b45616ff664b0fb5966a77435d24d1 : SUCCESS in 1h 23m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94a591552d7a4da2ae58af3568494bc1 : SUCCESS in 2h 00m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c0be94934b3c4cd0a60b7bc7e5536df0 : SUCCESS in 2h 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f71c2373b1384044b2057f6c7b8b18ee : FAILURE in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8da93f7c4daa321fe27e2676a72f7e6ec4bba4bb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-14 08:25:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d74fd862a6733efd83a7cf4ad2e2ef5b5383c2d5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-14 08:42:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc514d964eba0b138500e5eec2abe79a5b879861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 08:42:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3655aaeb02345badcae342069ca6608358e8dc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 09:19:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35608d4489c342ae9e678b376db1c031 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4f271c083b48e8a48343a1064b3fb9 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff06f6468e4740ecb46edb49d1bf60ce : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0280d8fb677846699f3b13a65bd5cabb : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8b0bd778f3480eac7eba501d052bff : TIMED_OUT in 31m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/619a39a78d38426b95cb1ce5f3198793 : SUCCESS in 4m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/82e1a9874c6a4a05a7be12cd4b931d7a : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb0de5f4ab764945a9607878cc5abb05 : SUCCESS in 17m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9421b9f781e24b3ab007276353db63b4 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ea0916f0283463fae078a9f36c204d1 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b324c86baf2241318c76229f6192a152 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a1afb6d2a9564ad29329a953567ba1c4 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ee4cf011d3c4b49bafc31f0b57c8585 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0790f618efcb4c369267a455bb722f36 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"69626e9748033a78e5cee8ba6f7e895effb6ffc6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 09:44:44.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeout in doc","accounts_in_message":[],"_revision_number":1},{"id":"ecd1b552f6b65436af0598a3bc57478a2ad9e037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:58:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61db89846f4a4c2bb674181907ff9870 : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3092bc568aa7438897e46d0c57e7793c : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e95fc984334716a6b0455965e0a527 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e637399fa282493e9808cd8137fce56c : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b0761348df4a2abdf87b6784b7111f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29b680bdef54195bf1521e4ca320501 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69202bbca0fb4afc9c14ec8e8735ed3a : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/355674d4177a48e48d573f9ed594be16 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/793e507e3a0e4c65b9368831405aabcc : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fdc99a86888d4e539a12c982eef116ef : SUCCESS in 4m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/237c292496d142338100f633651da6ac : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de5c0bea33e1437eb4916036f50f9468 : SUCCESS in 1h 29m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c9ca419fcbb4288b298e38e2469627c : SUCCESS in 1h 49m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d75b3349d16b48a59ab44f4341941739 : SUCCESS in 42m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e7b2fac0a2d49bbb4343ceffd460f0e : SUCCESS in 45m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6164d45b713f4b47bf9b1f9aae73664c : SUCCESS in 39m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/874622213eca475ca80ef53543f1ab69 : SUCCESS in 1h 06m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bf4939d30aae4a449d8481292307e7cc : SUCCESS in 1h 35m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d14363d306de44998eb70e4202972a2e : FAILURE in 1h 48m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/13f657952ad7443caacfe0c3b0ca9ffc : FAILURE in 2h 25m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4c5cf30f61454dc9a6820163c741f684 : FAILURE in 2h 23m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1ef4ce4f8ed5439899ba2fdcac4b2e38 : SUCCESS in 1h 55m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/db194c27f8514d32ae449e74efd3be2e : FAILURE in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbe929cd7043dd0e804932c387587b659a072905","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:59:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aab55d48d1fad4ddd8dc39ec11d8658663096531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 20:10:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18926f54c4284527853eb9514d1997a3 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000526532354402fa0484d6ef90655df : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12a7b7df9ed94545879f7b1f3bc7769f : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d93e6ee184a42d18f74d189c305ffe2 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d4583c67a04426aef3aa7a180e38af : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aee36757e0d4ff7bcfcf728c6db40f5 : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f00725a4db448008ce765e91994d17c : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd4b3f9554964916b1bbf0f6f49bbcc8 : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1d89e83c7bb4c92b02c665814fee9d7 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/721e1eb1b4114fafbb8c0efcedd6c394 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a0bf743aee74398a70d0688c4a676a7 : SUCCESS in 38m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a107b3644ce043eface4e78cb6ff19eb : SUCCESS in 41m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc732953c99f4998ae66f4832429bd4f : SUCCESS in 46m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f17237ab555459cb62153f4b37224f6 : FAILURE in 1h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d7f8cb3a0dec7052fb8e2de2becdb8cccded8a17","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 20:25:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5123956a0628225327ab6f8ece86f847a6942daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4d316fbbfa14c7ca0f41040fca35abe : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87fdcbec74ec4fdbb4b3728711a34a57 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0084da730754d4eaa9a8cc064198c0f : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d3eacba6c7e4e64a831225c095495af : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/915f5b484f85407098bb7884e2795e78 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae5827aab0648f7b31125561bd0ecae : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34643cbf9b4641878c4ed39554a0e072 : SUCCESS in 4m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/943185fb4c9e4d31adb196c676fb6cb0 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/050c62343fdb4e2c8c89c3193eebb31e : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/de13d71b5d5248ec8c65ca2ac4515dd9 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a96c24ce70b413da70f873057a77b09 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d40f67a23de54d9f961b613b6ee20d35 : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/374ca31302064e5190916a82296ec1d9 : SUCCESS in 1h 48m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a52dd63e9574336a39ed8d7716eee9e : SUCCESS in 40m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5a05d7f8e2f849549f6e643b7ed4097f : SUCCESS in 39m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edecf8fd710e4443b76ee9ffff123e04 : SUCCESS in 39m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e550431eb044cb5a937d700cd32f79f : SUCCESS in 58m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2cffe7b5e4d742ddb97fc70b128aca38 : SUCCESS in 2h 02m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2f6d4cfbf1a4507b361705797304b39 : FAILURE in 2h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8e340e31b6485a918170b270ed3d3b : SUCCESS in 1h 41m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e7f3159a9e58463c9431246789d49325 : FAILURE in 1h 48m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/33d4e83987ef47aea554e4b1c657383d : SUCCESS in 1h 59m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2fdf4226de96499890056c8f5ec79f88 : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b1995d93cc329d87417f7c9d440925beccdc23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:44:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3685d10ab3a9aa8bd629e502b4cc2b27b649da3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e65f593759b147adac397b3bf8333997 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c29851f92a44beb90da7ddf32a032d3 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c7c5469d2fa4acf9e6b1ce4d01a4723 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b78418be3194da6bdf500f2dad3bd86 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68586dd4e4b741259b5bac962fa59791 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46784784c6df4bb88da4bec6bbae756f : SUCCESS in 4m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/577e3b503f5a4f9c956efc4726dcbbfd : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88c90bc624a94689a55bda383a9b1cce : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/565f9e252c704b2bb3e482c7fc4d46d6 : SUCCESS in 1h 17m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b2c98ccfa6f4b42ad2603ba587d8b5f : SUCCESS in 2h 04m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad771ab5515640d9b1bc01b93d4e4aff : SUCCESS in 40m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/141b93f1471e4398b7400e10092b3346 : SUCCESS in 39m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b55c247c2eb49ba8134d86e0003b438 : SUCCESS in 40m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c5e2274598947f9a0bfe4083c125399 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1119af59f1a7801a1d87e2a8214f06f0d8067a14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:52:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be4441ba9d47d99037fdcc16621d16b2ad7b2b75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324cc047e2c1481a9253d2d4e28a8954 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d912db282254c7ca491601ce1287533 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64e313236efd224546775ac14f444900042aa394","revisions":{"64e313236efd224546775ac14f444900042aa394":{"kind":"REWORK","_number":1,"created":"2021-01-11 18:50:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/76/770176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/770176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/770176/1"}}},"commit":{"parents":[{"commit":"e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 07:22:35.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-11 18:50:55.000000000","tz":0},"subject":"Fix operating status for empty UDP pools","message":"Fix operating status for empty UDP pools\n\nFix empty UDP pools status: a UDP pool without members went OFFLINE\ninstead of ONLINE.\n\nThis commit changes the keepalived configuration with empty pools:\nthe configuration now contains a virtual_server and a comment about the\nexisting pool. This comment is used by the get_udp_listener_pool_status\nto detect that the pool exists and is not offline.\n\nStory 2007984\nTask 40610\n\nChange-Id: I30e23ca13d033d77c8ebdabbfdc7b54556a9466b\n(cherry picked from commit 2954370e386d54a88763128e2dbfb9c4b2bca96d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64e313236efd224546775ac14f444900042aa394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64e313236efd224546775ac14f444900042aa394"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
