)]}'
{"id":"openstack%2Foctavia~770444","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I6ce224d3f5413f1ba1f9972576ec51a89d249886","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I6ce224d3f5413f1ba1f9972576ec51a89d249886","subject":"Fix Ussuri requirements for Victoria grenade","status":"MERGED","created":"2021-01-12 16:16:35.000000000","updated":"2021-01-13 16:54:24.000000000","submitted":"2021-01-13 16:52:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770444","meta_rev_id":"66a52937ccbfecc2eed43b4084e9a97673df08d0","_number":770444,"virtual_id_number":770444,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 16:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-01-13 15:07:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-12 17:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-01-12 20:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2021-01-12 21:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-01-13 15:07:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2021-01-12 20:07:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":-2,"date":"2021-01-12 21:17:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 16:29:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-12 19:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 20:07:34.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-01-12 21:17:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-01-13 15:07:26.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"cec3de087ebfa1cb3b3a9e17334503aa0606281f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-12 16:16:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b67387970e49e504fced290032b5bc060555ffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-12 16:59:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd5dc7caca0b2f7c0ec56abe0d9515257bd1655e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-12 17:35:56.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e47d8dde6fdedf7739e620a52015d42c8a118d9c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-12 17:51:19.000000000","message":"Patch Set 3: Code-Review+1\n\nShip it!","accounts_in_message":[],"_revision_number":3},{"id":"8cd5ae1dd7cd8f90abe61462353b2e6152708dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 19:57:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/59d4706d3ea745bd8e1caa0fbbc7aea4 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d46aaeeaa6f9416dabe3d918380f6d75 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e77913e4ff6f4e9897c782eba740c1fd : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fec9e1767a4a678cd7980052507886 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a00c7395b9794f4db846f1ec871a1e3a : SUCCESS in 17m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2749fd4cbc14444828998ba9c6f99a7 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746bd7ecca5b41db83bcebd69636e802 : SUCCESS in 12m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4aedd04bc66a48c1ae5e174e758cd636 : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ce669e7e020e405d8ff2addb30616954 : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce161beb548642968d7d9289a40da1ca : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83f907c762c04cb4bb3d8e6688266b02 : SUCCESS in 1h 18m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a758593e70d464bbb9c2520c4f70958 : SUCCESS in 1h 48m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b8fca0dfe92a4f22a2574b45dac5a690 : SUCCESS in 41m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2fa7d6fd831d41a2a22a2f92cf826db6 : SUCCESS in 41m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76539c0fd7ab4e34acb64f56cfe1556c : SUCCESS in 48m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae928a4052dd439686a05957ef470ebd : SUCCESS in 1h 22m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4450c2ed272543108aec72422291138a : SUCCESS in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60ccb31eda03476a8c41baba4bdb52dd : FAILURE in 47m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a5f544465a84616bcd0e236967049c2 : SUCCESS in 2h 06m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40c8a7def2f8a58b2e13823366b82a964b520f04","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-12 20:07:34.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"a53d4ff69d9d185e6873194be5bf5258c5e91f7a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-12 20:07:42.000000000","message":"Patch Set 3: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3be0422aa8dac8158683906c72aa0a3432c63aaf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-12 20:55:56.000000000","message":"Patch Set 3:\n\nAnd my stable/victoria test patch shows this fixes the issue.\n\nhttps://review.opendev.org/c/openstack/octavia/+/770467\n\n2021-01-12 20:16:27.720 | Successfully built the amphora using the 01b9dcf39031182ac53cd9711ce09b12f200cbfd amphora-agent.","accounts_in_message":[],"_revision_number":3},{"id":"fddc7232d2ae9f8a661a5ae581d206c18b2dc466","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-12 21:17:49.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"a13dc84d83f3c72ddb227be8dc7bc7f4c304cbba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-13 15:07:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f8549659af518962edd860b9c327212b600cfa6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 15:08:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e5ced7260e006b4f0984149c2d1edf070f34c018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:52:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d70df63d40434701abaa8b84cb193afc : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623add94cd39486a9022c7f4562d6262 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a0d4e56c884ac5a32d4e9d5697514c : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2130c44a5a9c4a9a848313e1ee9d67a3 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d16b3e4b8144dcbbd0251fa16a8cc2f : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007b86b67e664488a38d59ac1dc2355d : SUCCESS in 13m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a1832fb5c49441badb75e76347b2c4a : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d66e09e3bbf49aa87890bc12e0605b3 : SUCCESS in 1h 11m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3d5aeec13414ae383f318fa2a44fb04 : SUCCESS in 1h 35m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ff33390d7b37456686f127d4e5defe66 : SUCCESS in 51m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/413e216704c94b9480dba03597672242 : SUCCESS in 38m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/733c758789bf4c5bbe43404098ef32c6 : SUCCESS in 50m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3ef2bed808e42c59ef6ae4da136fcea : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":3},{"id":"75ee0e3a7c176bf89680f76da03e346cae6e02f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:52:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"66a52937ccbfecc2eed43b4084e9a97673df08d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:54:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09aa1f9c4026441b932b965a2d19635d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01b9dcf39031182ac53cd9711ce09b12f200cbfd","revisions":{"93eabc2000ccf5f0bbff1716dccc21ec90c75360":{"kind":"REWORK","_number":1,"created":"2021-01-12 16:16:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/770444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/770444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/770444/1"}}},"commit":{"parents":[{"commit":"8816de41617c3dde88371496bad3e1ab8bb8324e","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8816de41617c3dde88371496bad3e1ab8bb8324e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 16:14:12.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 16:14:16.000000000","tz":60},"subject":"WIP Fix Ussuri requirements for Victoria grenade","message":"WIP Fix Ussuri requirements for Victoria grenade\n\nVictoria grenade job is broken because diskimage-builder is missing\nwhen deploying Ussuri.\nA previous commit that fixed l-c and requirements incorrectly removed it\n(Ia2b49128258b18653865622022e8877489f0087b)\n\nChange-Id: I6ce224d3f5413f1ba1f9972576ec51a89d249886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93eabc2000ccf5f0bbff1716dccc21ec90c75360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93eabc2000ccf5f0bbff1716dccc21ec90c75360"}]},"branch":"refs/heads/stable/ussuri"},"d2ac7d001ee1617cf45edb9c9f2fc90e1fd446c4":{"kind":"REWORK","_number":2,"created":"2021-01-12 16:59:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/770444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/770444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/770444/2"}}},"commit":{"parents":[{"commit":"8816de41617c3dde88371496bad3e1ab8bb8324e","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8816de41617c3dde88371496bad3e1ab8bb8324e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 16:14:12.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 16:58:46.000000000","tz":60},"subject":"WIP Fix Ussuri requirements for Victoria grenade","message":"WIP Fix Ussuri requirements for Victoria grenade\n\nVictoria grenade job is broken because diskimage-builder is missing\nwhen deploying Ussuri.\nA previous commit that fixed l-c and requirements incorrectly removed it\n(Ia2b49128258b18653865622022e8877489f0087b)\n\nChange-Id: I6ce224d3f5413f1ba1f9972576ec51a89d249886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2ac7d001ee1617cf45edb9c9f2fc90e1fd446c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2ac7d001ee1617cf45edb9c9f2fc90e1fd446c4"}]},"branch":"refs/heads/stable/ussuri"},"01b9dcf39031182ac53cd9711ce09b12f200cbfd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-12 17:35:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/44/770444/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/770444/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/770444/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/770444/3"}}},"commit":{"parents":[{"commit":"8816de41617c3dde88371496bad3e1ab8bb8324e","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8816de41617c3dde88371496bad3e1ab8bb8324e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 16:14:12.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-12 17:35:40.000000000","tz":60},"subject":"Fix Ussuri requirements for Victoria grenade","message":"Fix Ussuri requirements for Victoria grenade\n\nVictoria grenade job is broken because diskimage-builder is missing\nwhen deploying Ussuri.\nA previous commit that fixed l-c and requirements incorrectly removed it\n(Ia2b49128258b18653865622022e8877489f0087b)\n\nChange-Id: I6ce224d3f5413f1ba1f9972576ec51a89d249886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01b9dcf39031182ac53cd9711ce09b12f200cbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01b9dcf39031182ac53cd9711ce09b12f200cbfd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
