)]}'
{"id":"openstack%2Foctavia~770476","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I05488e91b33dfed891c0a896a86d8bd94bc4a918","project":"openstack/octavia","branch":"stable/victoria","topic":"oslo_lc_drop","hashtags":[],"change_id":"I05488e91b33dfed891c0a896a86d8bd94bc4a918","subject":"Drop lower constraints testing (stable Victoria)","status":"MERGED","created":"2021-01-12 21:21:41.000000000","updated":"2021-03-05 08:44:55.000000000","submitted":"2021-03-05 08:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770476-oslo_lc_drop","meta_rev_id":"bb45f483419c4c562c989f458d07fbe264d82381","_number":770476,"virtual_id_number":770476,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 08:41:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-14 08:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-28 10:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-01-28 10:38:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 23:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 08:43:38.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-28 10:38:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-03-03 16:51:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"13bededb2786361d8a4d91bdbe034093da9b5b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-12 21:21:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2cc48a3ef9bf50be513a430879f99328cf3eb55","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-12 21:22:20.000000000","message":"Patch Set 1:\n\nLet\u0027s see what others think...","accounts_in_message":[],"_revision_number":1},{"id":"5e2bf0b82b2eaf74d5aa026bfa59d57e164d3910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 23:49:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4a4a850ef364eea9f513cca991db66e : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e8569c5a1047a2be5a19628da25b38 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50302b6932404e5ea515cb5fd876f700 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7b491de114d41bda830be01e1ec4a00 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dea40d3b15f4892802ddf035c171f96 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2dc5626b1a34bb7a98adf42bb2e8fc7 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/492395bbe0704f5baef01a508f544790 : SUCCESS in 4m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a06437dad6f946e9b9a42cb0f586ea51 : SUCCESS in 4m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3498363ddead423b936ef1b4f34edd48 : SUCCESS in 10m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9f1cf33469ec4e46866a144e48ac13a9 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b11d6d202ae64daca8dc8674303c5f7c : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/253784acc897461ba5129ecdc9108d5b : SUCCESS in 1h 31m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/494036737b3f46f6888f3f5701b37c0e : FAILURE in 1h 41m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b13737769054c25b8082bca467d4d47 : SUCCESS in 39m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b34744124369430caaf29d5c07798349 : SUCCESS in 53m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75faebb474484e2aa86e9c0dac55e4fb : SUCCESS in 50m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb36ad053afe4e778fdcda7d1b1d3ce8 : FAILURE in 37m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cae67a57106246119e9bf92bbd28b911 : SUCCESS in 1h 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c04ed36fc29f4b1982ce20ca0158b984 : FAILURE in 2h 18m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b42aad2288ae4b16bed3ce4b062468b3 : SUCCESS in 2h 00m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5d880650644e446dbc62a303f5bdf9d8 : SUCCESS in 1h 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee5c68ef466d40aab6d8544a5665abd5 : SUCCESS in 2h 01m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/28b1e0fa53a14539894c3421338764e5 : FAILURE in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f59390b0799251fc587a253fe88c51c16f5473fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-13 17:42:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"725c92ed4cbaa9862546140cb921d29388a11112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:27:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/823064a10bfd45f29fdeb7dbbd475635 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9493de3f9bf44b90abfcc6aefe78cce5 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1ad5236b3e4c22bab71a80cdb4efb3 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4981fda01e594a8285755eb4c1d2dd2b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b61f8f206344cd4a94d900d67d16112 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71f7f9ed2ee4411b4c8d123da4247bd : SUCCESS in 21m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aefd69277134a58a86f0fb4bb379669 : SUCCESS in 11m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/33b74efbdd8f4929bac9210ee91dd7b3 : SUCCESS in 8m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bd33fd4caa2c4c3389e14175024dd9e9 : SUCCESS in 11m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4aa89bee51c44a09c1ce704d4c6ba53 : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d629ee742b0c459097b639a033293349 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6201f77f5854c558686d366a6ce3987 : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7149af0e1714093b9bb15fafdc5092e : SUCCESS in 1h 52m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0443053e9e041c292398d365f4d0f45 : SUCCESS in 44m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7f1f2c52ccd64692bfd4121a0069b496 : SUCCESS in 40m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cad9e9cdf8e40e0afc13e459d32b40e : SUCCESS in 43m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/80d36f700d464b0b8298d8fdfec137d0 : SUCCESS in 59m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9185a169f71f4bb6b8ec3c97d1d0cf5f : SUCCESS in 2h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48fba3b708e84e819fdb7e674c346b93 : FAILURE in 2h 11m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ec503cda9eb348a1b0ef7677fbcc9955 : SUCCESS in 1h 34m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/14e94145f7d144b68c7232b8aa9e1626 : FAILURE in 1h 55m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ad720f3ac9f54a5c9105668502ab8fb9 : FAILURE in 1h 55m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/753f0892c5244d8da4c263fe92b4c8e6 : FAILURE in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53f5c11c9e7dfd9bd985b9ef2ac173b9aaf76609","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-14 08:43:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94bba9fe6d2402eb5c8148c0e940015bf3055e98","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:38:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7796416e7cbe1c8947c3b2272d1c5f2fc66e9e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:39:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a3f15851852af9d5ddd0942603bb40e580d7444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:57:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2c6b8c7b8ee4a85b48d15f6f89a1505 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e244cace72e4d7dbccca627badcdffa : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b24100252bb9470680ced81d9261cfa6 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a9a95d34caf4416b3f42b2a1fb6b922 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b931ab38cdc4afda6e171dd858eb70c : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d00a6a6d42044382a52bdda818b504c0 : SUCCESS in 4m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f17f80e779054cc29dce97b69040f746 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f252a340022f451ebd72927909b4d412 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcb8f68ff92f4400810efede3426c8eb : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1bcf8c2aeb74f128558111d2fac89eb : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64c93f01c2d2430e8a64165b6de39638 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8fb21e5e7c1548c4982c880e86b24344 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d80fa87ccdea4e269a751c8dbd4771f6 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57bb0f9f90524b19b92385eddba1be84 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0030bcf2eccac6df2d7913a87df8fedac88b6836","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 21:59:44.000000000","message":"Patch Set 1:\n\nFailing in DIB 😭","accounts_in_message":[],"_revision_number":1},{"id":"32e3241cb7359e5c11e76f17f501df9c38dad9ee","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 16:51:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f30ae672521b67316a65907b5d61e05234885560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 19:54:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86aa2c10117c442cbc0cc50ddadd150a : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb37de906bf442578a9f953b9cc38003 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0111c61522940479b7447e5f4a4bb28 : FAILURE in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c9a6c6f20f84e16ad91921f1e814489 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/615afb6a24e7489995af432354504380 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1026cdf183b6453f8339a4155f1948d8 : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3767fcbce9b24147977ff4ef5fd72720 : SUCCESS in 7m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b62e8b92c372431782cf986b80a45037 : SUCCESS in 8m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/24f2b0115f3f482f842f5c302fccf57e : SUCCESS in 11m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f5d77519e1084bdf95bae299b73fc2c7 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b84cd0cd34fb43ecb67287b3d90f82a5 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19ddb8dd4e954690a228e3c6d94380df : SUCCESS in 1h 34m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ee0dc2fa7dc4fad8c64eb5dcf9e50ea : SUCCESS in 2h 13m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6fec8a1370184b82b17989d7dfa4950d : SUCCESS in 40m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/869df649b2c14721b292941ee03e931a : SUCCESS in 45m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fcbf6cbb7c24625815509eea7e58a39 : SUCCESS in 46m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fda0208c43fb4623aa4397d5de8415a0 : FAILURE in 39m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60b3ec8a670c48b190a6442442ba497d : SUCCESS in 1h 54m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/acd67c0e9288405c873f4e6d7c00b298 : FAILURE in 1h 52m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/91ebc4f75456487e85f63eaa4a0f7398 : SUCCESS in 1h 38m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e9e968638ac643d99617b83e44e0ca20 : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a858231cfe9f42f79ee83eac18fc875e : SUCCESS in 2h 12m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/207492f7c5324d8db7ccb5bcdaeafd7f : SUCCESS in 1h 49m 57s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"7dccda94e67f00593fa567c03fefa7ac8572dfd1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 04:13:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e41e8da42a8f87e6256e2e71c82a7ce2d0089f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b07a6463609b4761a632e5744da2997c : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd5bbb7bbbd84c5194f6df18dabe289d : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15d9049996c4dd6af30318dff3cb13c : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e942063649c84c7bafb7ebc8276c4b75 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a017d340cc340d38e81622f50bba014 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d0d060c36b44b193d071da1424d27f : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09234a64250e4590ad8fa354dfe50455 : SUCCESS in 6m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/24dce95d31094fb7a1f06595b7599658 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5f90ee5506e048919d01365bb96e0743 : SUCCESS in 9m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f85ffb565e2a4d1ea7ade083404e05d4 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a9e5344622641a49270af701db09360 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f059505eb40b4be0a67943293cdd8c3f : SUCCESS in 1h 09m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d41a6daf5284d188c6a6e7f04ef29d3 : SUCCESS in 1h 51m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e878cc1dd38a496ebc430d87c0150696 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1372d4e05e834f3ca584dfccc966bac9 : SUCCESS in 42m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10092c7a0a714704b8e8940ae7bdebd6 : SUCCESS in 42m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3a247a52c6864949be921969530e78a9 : SUCCESS in 1h 03m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b7b71e40f3964f12a50dfdff45066a1d : SUCCESS in 2h 19m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbb94c55b5c044f3b0ca10c860d4e610 : FAILURE in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac072cb259974d5dbe23c7800b5a57e0 : SUCCESS in 1h 44m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7af588b8fd4946c7a7905e20cca0a006 : SUCCESS in 2h 01m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/aae9590aa7c34955bc3f8adb493f994d : SUCCESS in 1h 48m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cac05c931395448d9c83e6c295bf605c : SUCCESS in 1h 58m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"870c31c2bc0d89d4906b635f7331740331cd000c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:38:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0ffdff850355244be2489bb93f54615c8308d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:41:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a8a084301734de582e351f90a85004a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8013b2481c4e1f8cc7f429047cdabc : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e991fcb3a14a41b78b291a5c5aaba654 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa90e9e87af495e8e89e2d69286f3eb : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8341dff5a642a7a32f13e987b67a75 : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645d6041a02f4013905b3e16f0ea8175 : SUCCESS in 6m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/193aadb237304ce5b56a4e508eb6a2e1 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82db8f4974c0477eb75e472f6efff41d : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67fefe0d03434f7798b0ea4aedddf083 : SUCCESS in 1h 14m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7f8381a7ac1441284f6d933b7435494 : SUCCESS in 1h 23m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c4906ff6a3a4407ae28eb3bc4566fab : SUCCESS in 45m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fe6cee32e5824ecf95148e8e89e7c7e5 : SUCCESS in 38m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c23e84d86e4b15ba88bf9ba3b048ed : SUCCESS in 40m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d646912ca5154e589e4b7e7a51718fbb : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0173183973670222b0efa33aa05830a9a582d75d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:41:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb45f483419c4c562c989f458d07fbe264d82381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:44:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9e82c9e13a460981a9466c01fe2555 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a323972eca3940a98b8f26aac1e46bc4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85875bcd318acee2af4f7ec314891892bd3a69ec","revisions":{"85875bcd318acee2af4f7ec314891892bd3a69ec":{"kind":"REWORK","_number":1,"created":"2021-01-12 21:21:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/76/770476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/770476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/770476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/770476/1"}}},"commit":{"parents":[{"commit":"e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c8aeffd6bb0bad0c0c0829b90d20b148f156bb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-12 21:16:51.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-12 21:16:51.000000000","tz":-300},"subject":"Drop lower constraints testing (stable Victoria)","message":"Drop lower constraints testing (stable Victoria)\n\nThis topic was discussed on the ML and QA team proposed to\nto drop testing lower-constraints [1].\n\nProposing to drop this test because the complexity and\nrecurring pain needed to maintain it now exceeds the\nbenefits provided by it on the stable branches.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: I05488e91b33dfed891c0a896a86d8bd94bc4a918\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85875bcd318acee2af4f7ec314891892bd3a69ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85875bcd318acee2af4f7ec314891892bd3a69ec"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
