)]}'
{"id":"openstack%2Foctavia~770937","triplet_id":"openstack%2Foctavia~master~Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2","subject":"Add support for building arm64 images","status":"NEW","created":"2021-01-15 14:03:59.000000000","updated":"2021-07-29 22:20:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac91a7e8b5f20c481cc8adf20ad7ae4886946e06","_number":770937,"virtual_id_number":770937,"owner":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-18 15:19:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"all":[{"value":1,"date":"2021-07-29 22:20:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"REVIEWER":[{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 14:05:56.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"CC"},{"updated":"2021-01-15 16:07:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-15 16:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 16:52:41.000000000","updated_by":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"reviewer":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"state":"REVIEWER"},{"updated":"2021-02-24 16:10:25.000000000","updated_by":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-07-29 22:20:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"9b747104d8a260d809656e005ccbd047feeb6af4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-01-15 14:03:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6914bac65dde2d8c64e6edbcf881a230d9928a5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-01-15 14:05:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0376d6ecc2ea1dce225b89469f7efa5e94446fa7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-15 14:05:56.000000000","message":"Patch Set 1:\n\nThanks for the patch!\nPlease note there is already on going efforts in Octavia to support ARM64-based deployments. See https://review.opendev.org/q/topic:%22octavia-arm64%22","accounts_in_message":[],"_revision_number":1},{"id":"b9a31f15fad58f7cc7277eb35e9a09f6f3ed8f45","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-01-15 14:10:51.000000000","message":"Patch Set 2:\n\nIf you are referring to this patch https://review.opendev.org/c/openstack/octavia/+/738096, I see it s been open for a while but not yet merged as it has too broader scope (devstack + diskimage create). I can change this one to the minimum necessary so that we can merge it faster?","accounts_in_message":[],"_revision_number":2},{"id":"61a5c99bb279fe01a6b5e0c39d7206ae87a95125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-01-15 14:14:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"87267bac3e06faa28a854be9e2fa80f799a3acbc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-15 14:15:37.000000000","message":"Patch Set 2:\n\nYes, I\u0027m referring to that patch. The patch covers diskimage create tool just like this patch of yours as well as devstack as you correctly mentioned. The reason for such is because we want to make sure we merge functional code tested in upstream CI and that led to another patch -- https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/747629","accounts_in_message":[],"_revision_number":2},{"id":"d9d2d1ec3c950026db3071065d16cbd22d3896eb","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-01-15 14:21:16.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Yes, I\u0027m referring to that patch. The patch covers diskimage create tool just like this patch of yours as well as devstack as you correctly mentioned. The reason for such is because we want to make sure we merge functional code tested in upstream CI and that led to another patch -- https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/747629\n\nI suggest to not use aarch64, and use arm64, as it can be reused for Ubuntu images too. Also, if you use arm64, there is no need to change the package installs file.\n\nAnd dividing the patches in smaller atomic ones can lead to a faster merge.\n\nThank you.","accounts_in_message":[],"_revision_number":3},{"id":"519c647af9894a4f8dc5882f08e040321e61463d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 16:32:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0cf81deb5ef4794b287e2eb8246c180 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b46987856ba4dc98e7fe1cb30a98880 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4bc0b77cb074e5087c6aa51ae6356ef : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2faf952f0bc4809a3ad4f95c8f5e1b9 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53a60fa8e884480afedad78de632654 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de7dc6825b9e46c1aa5218dc43b46f55 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf7be65700742cb9f5452bb9ff36416 : SUCCESS in 13m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7211470d219c493292047450cf48921f : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0dcbb7ed59c549aaa6438664ece020a2 : SUCCESS in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/02a1493fa13e40f089fce6306b7bb049 : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d754d8cc17a4cacbafbd5b6be0b72fb : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed897f06f8104d0482d748790a27d827 : SUCCESS in 1h 23m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ab583fe72b94caea1b1283bd110ad4c : SUCCESS in 1h 39m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3f1f0a179eb843e1a3169257520e8124 : SUCCESS in 42m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/02f82ac3f012473fbdf69f070b55c86a : SUCCESS in 39m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15f7a0fe2cdd4257aeaa443e03f13cb9 : SUCCESS in 45m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29f2c32cbe904071938661888c224f73 : SUCCESS in 1h 01m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2bd0affa525145c8b48dddef29b0a966 : SUCCESS in 1h 54m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d9a96e8bfc6a4b30988825055963c914 : FAILURE in 1h 56m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a7070a215a04406ea93ce9f65ab32304 : SUCCESS in 1h 58m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd06e25be60f4164b77d141a64ddc869 : SUCCESS in 1h 56m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4cdcb054181b4d409f58dbc239ccbdfe : SUCCESS in 2h 14m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a09eb4c486fe475fb4b767ea1b0590df : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55fe4e4073956b52bebed696d1b514d11f3a4dc7","author":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"},"date":"2021-01-20 16:52:41.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b0dcaeb5a813cca91fdcab99fa2b41078ef90072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-02-18 12:47:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ad8a13b29990a7d4358f0023add68ff2c1e91f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-02-18 12:48:45.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4496f200052f8309dfce1a9b38b69f0f614af8dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:19:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7b2bb9467ee4e1c93978a4d043c7c9e : FAILURE in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afcfa9f21f9e4ee28cb2e5bf13cc780d : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a98720e1a94bf38fdb464fd402c3dd : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c2257d0b1ba4262825abec37383141f : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102658a3f1ef4f59bfbe63b08efa60a4 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4e69b46e6974c1dbfc612bb63fea9a5 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab01da07c184739b385be1130e2ed65 : FAILURE in 9m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/77c68d00c9774c65b1cca1c959ca5853 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/491b6166748e4187abc989b32dd65786 : SUCCESS in 10m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1892603d1de54baf97ed8c9ea3bf1ffc : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef15a1bdaa0848c0a25e745eb7fb428d : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee5c6f5381374e09a7748114527ce086 : SUCCESS in 1h 26m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8de668f8a16a4c3da9c6c9825460293d : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e124ad90dbf46afae3b42b47ab9a6cf : SUCCESS in 42m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9d6e6745966c49d0a20d3b9db42c8883 : SUCCESS in 1h 14m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2959f6cab4f44355be399046daa1efa3 : SUCCESS in 41m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/598081c839ea42f3ac192e1bce4e0c0c : SUCCESS in 49m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2df431200874de7b4d02f33c410f4a1 : FAILURE in 2h 06m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ef9a17130764633a5abc712df620419 : FAILURE in 2h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce3e623b57134c4c83a9b1a5f22f854d : SUCCESS in 1h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45fde5a11fef46bca6bd7e027b12c484 : SUCCESS in 1h 52m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b306169b4db4dc69e047729e4a95602 : SUCCESS in 1h 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/79d97794117942658358e9f459c479a1 : FAILURE in 44m 23s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/install/install-ubuntu.rst","accounts_in_message":[],"_revision_number":5},{"id":"d8ba9c6a86f14aa7bb60d1f96c2b5740b6b42315","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-02-24 16:10:25.000000000","message":"Patch Set 5:\n\nHello Marcin,\n\nwould be great if you could take a look at this patch.\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"24e04e4e49bc52804be84c9c86a32802e19ab3b1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-02-24 18:02:28.000000000","message":"Patch Set 5:\n\nCheck does it builds CentOS and Ubuntu images.\n\nDebian based systems use \u0027arm64\u0027 as architecture name while Fedora based ones use \u0027aarch64\u0027.","accounts_in_message":[],"_revision_number":5},{"id":"bf856386c5e9fd219c5b2d9a427d27fea7abf878","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2021-02-24 18:40:59.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"394ea7741c1f13fd6da9ed679cd7dbe3f726a7b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:02:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86f6ed73e19245d48f7d9df2ce6ca4fb : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/290dddc552f84177a44d4c06192dcfba : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3e4801483c4546aed03680a7a7c5ba : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/152ccfaa2a72410b96941bd0e1759770 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/077a1cfe8e2a4c7ea5d9898814559dbb : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8ac4050995340f19c828e93a6b6bac7 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39637b51de5043c7808cc1e83ac7c42a : SUCCESS in 15m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b82a6321a16a4e5ca9dd016d1feac8d5 : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/101da9af80c0446fac3d7548efc61fa7 : SUCCESS in 10m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/77c72df5b9fe42ebb54052f1a20d9c8b : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8475652d2984da589c8b61b360ecc7d : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4b517b16ed5491bad2f718474ea9dac : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3355497f5f564b379b5a8d7b4785f180 : FAILURE in 1h 51m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e83e26fa70c441f8a6debf98d891b04 : SUCCESS in 50m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ebbb93d40f11485a8db4b0fb580d9ecc : SUCCESS in 39m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f0030b20c3e4c32bc772b4f39a1724b : SUCCESS in 41m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5ce77ce9b1441f69b54256e9fd160e5 : SUCCESS in 1h 02m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/66611f70713041ebac032befe014c0b3 : SUCCESS in 1h 42m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0da2779ef6845e4925d2c2556c4e18b : FAILURE in 2h 19m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c675af877de44f7f8dfd832dd89b653d : SUCCESS in 1h 42m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/14fbc49947e3442ca577e44539164964 : SUCCESS in 2h 25m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f9bfaefb5caa428e9c487fd3204bdaff : SUCCESS in 2h 11m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/21128eea68e242c8be122d468d412623 : FAILURE in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c16c9742966f50e4d53a48d300f543bb98ccef7","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:20:51.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e8b9d9601503b8c4d89541cd4c2ac3919b09fe08","revisions":{"bd724c165852f8ad9b122694b7d5d08a25c436f2":{"kind":"REWORK","_number":1,"created":"2021-01-15 14:03:59.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/37/770937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/770937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/770937/1"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:01:05.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:03:57.000000000","tz":120},"subject":"Add support for building arm64 images","message":"Add support for building arm64 images\n\nDiskimage-builder does not support armhf architecture for building\nCentOS images, as it cannot download the base image because the\nbase image URL is constructed using the architecture value.\n\nChange-Id: Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd724c165852f8ad9b122694b7d5d08a25c436f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd724c165852f8ad9b122694b7d5d08a25c436f2"}]},"branch":"refs/heads/master"},"a113dc95482b829decf0cc1e0b01a3b53ffb1ffb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-15 14:05:28.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/37/770937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/770937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/770937/2"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:01:05.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:05:10.000000000","tz":120},"subject":"Add support for building CentOS arm64 images","message":"Add support for building CentOS arm64 images\n\nDiskimage-builder does not support armhf architecture for building\nCentOS images, as it cannot download the base image because the\nbase image URL is constructed using the architecture value mapping to\naarch64 and the existent armhf is not mapped to aarch64.\n\nChange-Id: Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a113dc95482b829decf0cc1e0b01a3b53ffb1ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a113dc95482b829decf0cc1e0b01a3b53ffb1ffb"}]},"branch":"refs/heads/master"},"c757fa0447d2f20f190dbbf60fc91c057106bb5c":{"kind":"REWORK","_number":3,"created":"2021-01-15 14:14:36.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/37/770937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/770937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/770937/3"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:01:05.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:14:18.000000000","tz":120},"subject":"Add support for building CentOS arm64 images","message":"Add support for building CentOS arm64 images\n\nDiskimage-builder does not support armhf architecture for building\nCentOS images, as it cannot download the base image because the\nbase image URL is constructed using the architecture value mapping to\naarch64 and the existent armhf is not mapped to aarch64.\n\nChange-Id: Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c757fa0447d2f20f190dbbf60fc91c057106bb5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c757fa0447d2f20f190dbbf60fc91c057106bb5c"}]},"branch":"refs/heads/master"},"f4469494d22697e9854f2ed358c9a1392672b3bb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-18 12:47:35.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/37/770937/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/770937/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/770937/4"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:01:05.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-02-18 12:46:45.000000000","tz":120},"subject":"Add support for building arm64 images","message":"Add support for building arm64 images\n\nAdded option for the architecture (-a) flag so that ARM64 images\ncan be built.\n\nChange-Id: Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4469494d22697e9854f2ed358c9a1392672b3bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4469494d22697e9854f2ed358c9a1392672b3bb"}]},"branch":"refs/heads/master"},"e8b9d9601503b8c4d89541cd4c2ac3919b09fe08":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-18 12:48:45.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/37/770937/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/770937/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/770937/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/770937/5"}}},"commit":{"parents":[{"commit":"b62d9f02147195cbf07de5b2c7a7504bb764d7a8","subject":"Merge \"Fix operating status for empty UDP pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b62d9f02147195cbf07de5b2c7a7504bb764d7a8"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-01-15 14:01:05.000000000","tz":120},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2021-02-18 12:48:19.000000000","tz":120},"subject":"Add support for building arm64 images","message":"Add support for building arm64 images\n\nAdded option `arm64` for the architecture (-a) flag so that ARM64\nimages can be built.\n\nChange-Id: Ie2c1bacc920d4d8639d1ceb7251ab5af1ecd26f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8b9d9601503b8c4d89541cd4c2ac3919b09fe08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8b9d9601503b8c4d89541cd4c2ac3919b09fe08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":3185,"name":"Alessandro Pilotti","email":"apilotti@cloudbasesolutions.com","username":"alexpilotti"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
