)]}'
{"id":"openstack%2Foctavia~771759","triplet_id":"openstack%2Foctavia~master~I1c5cb987945a7a465bfecfda399dfe93fc1b76bb","project":"openstack/octavia","branch":"master","topic":"sctp_proto_validation","hashtags":[],"change_id":"I1c5cb987945a7a465bfecfda399dfe93fc1b76bb","subject":"Validate listener protocol in amphora driver","status":"MERGED","created":"2021-01-21 08:10:58.000000000","updated":"2021-01-26 10:14:48.000000000","submitted":"2021-01-26 10:12:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771759-sctp_proto_validation","meta_rev_id":"f12975fc8f783b67cd2ba1d48a44e71bfa0dbab8","_number":771759,"virtual_id_number":771759,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 10:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-26 07:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-21 09:42:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-21 14:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-26 08:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-21 09:42:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 08:31:04.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-01-21 11:57:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 14:58:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-26 07:54:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"771d754e1ba861b5598ef764292d4311b79eadc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-21 08:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ca66c09068bc0004d484ae03d20d05284364436","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-21 08:13:43.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nNeeded in victoria","accounts_in_message":[],"_revision_number":1},{"id":"201772a1b6a5df6d38994c8b640f12da348773a8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-21 08:31:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nSorry to be a pain 😄","accounts_in_message":[],"_revision_number":1},{"id":"f64b04d2fc5acd303f272188147f313e6a9430d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-21 09:37:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b6419f622ca6a0951905d4240d91bc6402f7b36","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-21 09:38:24.000000000","message":"Patch Set 1:\n\n(3 comments)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e Sorry to be a pain 😄\n\nI guess things will not change in 2021...","accounts_in_message":[],"_revision_number":1},{"id":"732d3e2741129a08c2bb6c23fd6a4c539a94a322","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-21 09:42:52.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\n\u003e \u003e Sorry to be a pain 😄\n\u003e I guess things will not change in 2021...\n\nI want you to know I read this and that you are not wrong 😎","accounts_in_message":[],"_revision_number":2},{"id":"2c1c5f0e19db6088406f2fd94db5f33b9f3ac8f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:57:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5508927eafd4dc48f0a5974fd5c48dc : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3afeb1b9a8234ed58ff372a22d173ea6 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f38ccfff96549e4881c9216ba45fcad : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9615df635af24d0c87aaf91f5e49ef6e : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0795b6c7a80b42a9afea8499b75d8374 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf41791bdc4944488696e5d783748615 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f50b14bc7d4693b0aaeddfaed62b84 : SUCCESS in 11m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/659c81a586c7447db3bb7bfdce2b2251 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/981fe1c1088a4443af9f18590231b61b : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2818a2626b304aada37f009c39314e7c : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/294362c590d74d07898fe8fd2bd26ab9 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0cc3873c215e494d97b7e1411d4f87a8 : SUCCESS in 1h 13m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab73c058299949deb62e397e47113dd6 : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a642a68d068450cafa16f3e967dfa79 : SUCCESS in 40m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/567f095f36d649b3b31e83a267bcf308 : SUCCESS in 46m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c1fb4fbc7b14740bca0e16a731a8c98 : SUCCESS in 50m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a9e39570fa834b5cb6d384f9b184d430 : SUCCESS in 47m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d48877c180f445dc91eb5bd477db4014 : SUCCESS in 2h 02m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/476afcb7d72d4bb6a323d4cd320d0e86 : FAILURE in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2fd1674b4c848f39a8a5bb655960616 : SUCCESS in 2h 03m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8027669841b5410094abd8ebe3ba42de : FAILURE in 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9d280b92aadc4ec38448806ae62c7038 : SUCCESS in 2h 14m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b0934017e4104e6ab54155c919584ffd : FAILURE in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf18ba9074cf9a1aa0ba340809f4abc48f014b16","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-21 14:58:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1b7111eafd92ff9bc2bea7d658a4144650c030d9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-26 07:54:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nSmall question and we can merge this.","accounts_in_message":[],"_revision_number":2},{"id":"835e601f7d3cbecb7fac36847cb64e70d51af6db","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-26 08:03:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3d1440e7e6d54352c9b904da698c86ae184d5e7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-26 08:13:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61762fcf7aecbeffe6e5085b5eac37b3bb6b7ea1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-26 08:25:35.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f1a2ae16092a3011cf01189637882df213f1340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 08:25:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24025e5827a2e30970eff25e4b3f67835498137e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:12:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe4ad83b30bb4ca08485b31bc9373abd : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391a116cc4cd41c280d435f92c326a00 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ed07c20343f411fb882f870d96d75af : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e15946bb21a491c9b8383705fb29903 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d92da81ef94d1b89f717d450a0522d : SUCCESS in 12m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a126a0c185464226b151255c89311c33 : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c78b6f2da1ea47e4959b432781c8b715 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a50f053d06464d1c875a930a7aa8b5c8 : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fd308500c6142bb83be010056bc0b81 : SUCCESS in 1h 45m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1b58d0ff87c43be9268a9907f658adb : SUCCESS in 40m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/011899d43580486381ac11dafe2cd99a : SUCCESS in 39m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0427db3b68d4da592fc455030b1a2ea : SUCCESS in 53m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76d87b87e5234392a04520316732f4af : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0ef4734e2ad81bf5ffd28884ffe817d06d716fa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:12:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f12975fc8f783b67cd2ba1d48a44e71bfa0dbab8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:14:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4881068bc8604ae6aa21251226daecc3 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33","revisions":{"0031d9c6544da55ef4af50ec3269844bc8e36b0d":{"kind":"REWORK","_number":1,"created":"2021-01-21 08:10:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/771759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/771759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/771759/1"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-19 17:16:53.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-21 07:43:58.000000000","tz":60},"subject":"Validate listener protocol in amphora driver","message":"Validate listener protocol in amphora driver\n\nValidate that the amphora driver supports the listener protocol when\nreceving a listener_create request.\n\nIt returns an UnsupportedOptionError exception to the user if the\namphora driver doesn\u0027t support a protocol that is defined in the API\n(ex: SCTP is supported in the API in the Victoria release, but not in\nthe amphora driver).\n\nStory: 2008545\nTask: 41647\n\nChange-Id: I1c5cb987945a7a465bfecfda399dfe93fc1b76bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0031d9c6544da55ef4af50ec3269844bc8e36b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0031d9c6544da55ef4af50ec3269844bc8e36b0d"}]},"branch":"refs/heads/master"},"fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33":{"kind":"REWORK","_number":2,"created":"2021-01-21 09:37:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/771759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/771759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/771759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/771759/2"}}},"commit":{"parents":[{"commit":"a4aa03d3bc98eb27cc353140cd998a623baa505f","subject":"Fix the unit test tree structure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4aa03d3bc98eb27cc353140cd998a623baa505f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-19 17:16:53.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-21 09:00:47.000000000","tz":60},"subject":"Validate listener protocol in amphora driver","message":"Validate listener protocol in amphora driver\n\nValidate that the amphora driver supports the listener protocol when\nreceving a listener_create request.\n\nIt returns an UnsupportedOptionError exception to the user if the\namphora driver doesn\u0027t support a protocol that is defined in the API\n(ex: SCTP is supported in the API in the Victoria release, but not in\nthe amphora driver).\n\nStory: 2008545\nTask: 41647\n\nChange-Id: I1c5cb987945a7a465bfecfda399dfe93fc1b76bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
