)]}'
{"id":"openstack%2Foctavia~773598","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I64a7072b2c4e1a12b5d908647f1eddcad7ec3b90","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I64a7072b2c4e1a12b5d908647f1eddcad7ec3b90","subject":"Fix setting None in tls_versions and tls_ciphers in pools","status":"MERGED","created":"2021-02-02 08:02:39.000000000","updated":"2021-03-05 13:59:26.000000000","submitted":"2021-03-05 13:57:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773598","meta_rev_id":"91705d664841f6136e9b12bb289d7e7f1577304f","_number":773598,"virtual_id_number":773598,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2021-03-05 09:21:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2021-02-26 18:11:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-03-04 00:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-04 00:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2021-03-05 09:21:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}],"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 10:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 08:57:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"CC"},{"updated":"2021-02-26 18:11:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-04 00:27:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 09:21:03.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-03-05 09:22:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"CC"}],"messages":[{"id":"b238de7f615d8ad826dfe4267d1e15e79bf7d94c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:02:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d7ea58be36ff0c19c9297c83cc27584a7dd3584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 10:35:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3064d7f7607c4e92acda3e0018b42fff : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8189b55ac246402e91377bfc430d3f1c : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16501172ee6d4cb3b2380b6e139e07d3 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c423dee0d5f944bcb8a9dcac345eb106 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bebed90a49234bd493285304c1625afe : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702853691bbf4d109a44a116a409e2bb : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ca485a387034e8ea9343d472e04f268 : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ceb2f484d8d64d1abe62d4c24852812c : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/94e50ac1b4ed4d528394f4e6dffee7f5 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20987e72f9e84ff9a97c305d5d468944 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da15973d6d584797b4f29b7e5d403fc5 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a7fa2dd2cde42f5af4cb514bcedb163 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bee7dddc63c8417280166af7ee3479ae : FAILURE in 2h 16m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0481f554f7ce4393b248e65ba62c31e9 : SUCCESS in 57m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0f06553edc784e4aaae3f0ad595e23b3 : SUCCESS in 54m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b0b8cbfd5884693a37f57c33f000c1f : SUCCESS in 33m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c88799c54ec1485ebab887c0ae2e3cfa : FAILURE in 35m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e5f1a6ae35e41e186b4105259694b86 : SUCCESS in 1h 58m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adc2a128f2974fdf80838916903f0b6b : FAILURE in 2h 25m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3ea2ed10c1440b19d93b529936a7598 : SUCCESS in 1h 50m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b170acb89d7342ba8562ddf85fe414d2 : SUCCESS in 1h 49m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/13e9ca7f8b7f44a8b92aaa365e7d99c5 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8cab10e8432c450a8a3cc8658ea9acdf : FAILURE in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3baf2157141e4711d20870f2d95b49bd3695d42c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:57:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc1f70c1ca2f3e17175755bfda5f7d2c1d6582ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:42:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa52cc58d71f4591a07d0344333c9b53 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5edc13047898469bb12b7d33dc370d63 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ef3abcf09e4419b76aaa55c7ab7a94 : FAILURE in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f0b1d7b2133423482c0aeb3b9732678 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a750c66ddc5946189c2c330819967d0b : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423b24f257c941d7be6f831e74140ccf : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9cbbfbc0b28488499fa6cee5169562b : SUCCESS in 5m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c2fecc22d50d48b2bc1e2388a7d3047d : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/caad3297ba3d4d9babfec2f1ca3c3296 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b733a1b13b474881ab17ea0c35b7f091 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40eb98217f654ab88c326fbed1ad0671 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bb797887190d4ce48b13d12132ee8986 : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/029c7116fa434a038f890e2854412013 : SUCCESS in 1h 57m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ba3a5291ec145999b2aced2ac2f59f8 : SUCCESS in 44m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb860ec13ee848e384d92e71bcc0f9cd : SUCCESS in 52m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d393e20684d84958a859270afb030ad6 : SUCCESS in 59m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e20e0459c464afc9839527b7f0905ff : FAILURE in 31m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e297bd33277455e91080ed3449e61de : SUCCESS in 1h 52m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa7ec236c59740639dd7f05c49a3c3dd : FAILURE in 1h 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6164836c487c44fbb6c14f00156a1183 : SUCCESS in 1h 47m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23228673dd58441d9dbe5b7b69c652a1 : SUCCESS in 2h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e8f6bf74eba84065a8a3e67258e0ec0b : SUCCESS in 1h 59m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dedb312d047f42e8b47ffa1263b5c81f : SUCCESS in 1h 49m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/policy.py","accounts_in_message":[],"_revision_number":1},{"id":"155bef0a5f7cdc0950c87fa35267a3e529d779a9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:11:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5e26ab25277251967228c47b0715abd2298308e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:27:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4128cb75dab8b91c985d92712d6e4cbf00ff4b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:46:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/648dd865218d4e21ab18e79fe12e9250 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6221dddf205745d09ce6c0b66f428d07 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fafb85609d4104984a969e4d3e5fbd : FAILURE in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c067242369cc4b8e9b0b10fb87eca8f1 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/158ebdaa346747178bc34aff1a64b1c4 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83780e8a893543f7831f9624823a7aa4 : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be5def8e7e24390a8d057be8d3c61f8 : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c0ed7081c07c46b8abb45fc34578a768 : SUCCESS in 7m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e1f951967af747a7a54ca39576d19064 : SUCCESS in 10m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d8f7c72e495d4c0b87fec0096399fdd4 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd0624c158bf447a8e78da9f77893224 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f9da133aee5a4df69c996999b17f068f : SUCCESS in 1h 16m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3fe27c1eb334d3d97291bd126928a9a : SUCCESS in 1h 54m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ac2df9cd27348b58fcb170d6d829065 : SUCCESS in 47m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1d1991224da94151a73d14d41b494486 : SUCCESS in 42m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b7ac9b248924986ad29df97c6180dce : SUCCESS in 42m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/787314660f8440d4adf69b32da3c46e6 : FAILURE in 32m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15fbf9a025e2420cafa2740cc58552e0 : SUCCESS in 2h 16m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/321ad4d1b0e7424da1372e816a9311f6 : FAILURE in 1h 54m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e3d1da42b00545b4808ef0cf7cf1f9b9 : SUCCESS in 1h 55m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39969edc8a3243ad9712e3c1d10056f1 : SUCCESS in 1h 28m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/30c16ac1775f44b590df6bdf2140fb44 : SUCCESS in 1h 50m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c814e7d2fc8b4cf4b5e8357f88c3e233 : SUCCESS in 1h 41m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"2521d76eb4e8ef65ba2281123d1fae9bd7d1f57a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 04:07:42.000000000","message":"Patch Set 1:\n\nRecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e041f81a595ce36a02c9a5a1bb05e7ed40b6fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:17:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04727f2efbd3453f8d82f38cdc6700b8 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2740f8e7d9f44c9995470207b3ddebc : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d10e720f5847f3b2b11c86ab5befab : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fb8105d51a54f84866f323170d2022a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f96bb820041c4b969f54a0efdd460085 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfa9a38bbe349ea82b49e2b12e180de : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f5678e5e3074b2d89f33e0c2fc18172 : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/956cca7aef1f4f0e9f602a33dec701c3 : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/544e5154fdd649788cd7c81bcd4b760a : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0cd0c44851d4e02967bf6736054e9d4 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7f8580db4f246299800a1820ac90d58 : FAILURE in 10m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18cef20a0ee949f9886c1a4fbad9db02 : SUCCESS in 1h 10m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b85410821c83441d89aeb1bd6bab5c1a : SUCCESS in 1h 55m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/188b78ad14e9424e970db19858a7ca2f : SUCCESS in 42m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bc5c42e1222546398e195f64eaaae2bf : SUCCESS in 38m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c740869725a948c28ad0b358270a81b2 : SUCCESS in 41m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4e845da0c1e404a9f07cac9671b3b2d : SUCCESS in 1h 19m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fd45903b2ef342c395de1053c7fe6319 : SUCCESS in 1h 47m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f2174aade3e4f97a264186e0f7471be : FAILURE in 2h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/534baeccd2e94551a6def4135db6fb6a : SUCCESS in 1h 41m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8891e96ff2d04fd6861c7d0745e86b79 : SUCCESS in 1h 39m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3737679e06b5426ba22658c289803aa9 : SUCCESS in 2h 05m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8265b9c94f9748b480e91c8169b14cb3 : SUCCESS in 2h 06m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"281d5b5290e74aa4d4fe7293a954d9324bab0434","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-05 09:21:03.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"1338e0eca0e0f2ab7f49340733880fa9ac693357","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-05 09:21:11.000000000","message":"Removed Backport-Candidate+1 by Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"3ed8eb8582f96fe69751dd8b5cbcbba289fac0fa","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-05 09:21:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9202d9ae63f241d0075f3bf37bacc64691d4e9fb","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:22:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67da9f7f04bfc1df5a81e51e468a7d99e62a06b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c8e5b3b7df422284a4f8b295354e9a : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2efa7574ad4751ad76ca50130f5c34 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27cab6b4a4bb443aa0eff7de85c50246 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e230f4a94224eeb8f60d5196d94782a : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288e2bfa814b43518bf5b6b492dcb278 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/732a1c44789641469a5bc0428a0a10c1 : SUCCESS in 6m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/af86bb847dd64824870194c1b2592df3 : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9b034dce283a4af1bdce3ea3d43b59dc : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81595e7cacf24259832cbcf5ea31b91a : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d631879a68a54169af4127a5b4f87feb : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2dc064aa33434250a6215894cac23b6e : SUCCESS in 1h 00m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fac22712101a465d9d617c0e1db35bfc : SUCCESS in 2h 11m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a72985baa55c4d268b44ac120738807a : SUCCESS in 1h 00m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a145b25a12dd4589907fe537f18ab1f0 : SUCCESS in 39m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92ab5922494f45c8b2ad6c3792ee121c : SUCCESS in 43m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72cd997493b446e58d30fa67004e72e3 : SUCCESS in 1h 22m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/42814992dc8c447db4a2cf0c547cae42 : SUCCESS in 1h 32m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1054202e51334ea6a88290c5accd2f1c : FAILURE in 2h 19m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d2073e9993d64d08a22897a125bce560 : SUCCESS in 2h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcb773dc655e4770ad7cd7aa0e57a7fc : SUCCESS in 1h 58m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7c5edefd412b4504a3ac0681276bdf94 : SUCCESS in 2h 31m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f415e7019ced446887d311e219034e97 : SUCCESS in 2h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9751fa7203336f7f8ba72eb41e002ed5980b37a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:58:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f837a2bc467019dcfb759b54f4902d7f84a1972b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:56:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1d21514bc349059b67ab65dfcdcdff : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5b0183790f64d1784410a488e302725 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b73939273514db68e843bc0de178675 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb32daee5fa24858b38e4db293430ebb : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c40b02a41ae84b18927ea48eb5c89eae : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/26750b16e68246b7964631941c454bf4 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bca5d5bcad7417eb1aa058a966468e6 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c440c094bd004eeaa3b0e3cf4a9bc4c7 : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/158d8349653e4d6d8c8df9b455c3e646 : SUCCESS in 1h 55m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07fe81805c704b01b58fd42bd0a468bc : SUCCESS in 46m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a93cbdf24b414292b943bde582cd2cc3 : SUCCESS in 55m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b89418b4d1fc4472a93d1f092814db91 : SUCCESS in 45m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95e1e11661c84307a17ee0c51c69462c : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fff9630315059aa3c2a458e5a773104731aa32eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:57:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91705d664841f6136e9b12bb289d7e7f1577304f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:59:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522434d63dc447919dd477133bc2c24e : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6b1fd4b0ae649b0a8c318f834ed635b : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"550c1341f9f059f1792944c0060bd7a0cd5ca448","revisions":{"550c1341f9f059f1792944c0060bd7a0cd5ca448":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:02:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/773598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/773598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/773598/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-18 13:43:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:00:46.000000000","tz":60},"subject":"Fix setting None in tls_versions and tls_ciphers in pools","message":"Fix setting None in tls_versions and tls_ciphers in pools\n\nSetting None in tls_versions and tls_ciphers should reset their value\nto the default value\n\nStory 2008367\nTask 41276\n\nChange-Id: I64a7072b2c4e1a12b5d908647f1eddcad7ec3b90\n(cherry picked from commit 8a207a7e036538abe68db38539c2c785537171dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/550c1341f9f059f1792944c0060bd7a0cd5ca448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/550c1341f9f059f1792944c0060bd7a0cd5ca448"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
