)]}'
{"id":"openstack%2Foctavia~773600","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615","subject":"Fix nf_conntrack_buckets sysctl in Amphora","status":"MERGED","created":"2021-02-02 08:14:10.000000000","updated":"2021-02-26 22:10:19.000000000","submitted":"2021-02-26 22:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773600","meta_rev_id":"a4ee2f4c77bf2af6170a6735c30e1fab90bf67f9","_number":773600,"virtual_id_number":773600,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 22:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-24 09:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-02-26 18:10:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-02-26 18:10:36.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 09:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 09:27:01.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-26 18:10:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"d6b1b1faff1f1056fa069d91a070a9185b4e67fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:14:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f86965e6a9c6cbfea7667252ea4d265b3c093e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 09:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ee3807ba778413eb7ea4de67b95a1dc : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35f0eb8e08f74dd6921c28beffce9f32 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9882e444a02043b99b02b994e0c3d60d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c6a49e9bbb74d3da27833543f201071 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eaf041963f94d6890817831178612b3 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1900aa6f70984f349b3aa1b78a8b2251 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/457880b97f5549d3a177e66f9d62e312 : SUCCESS in 4m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5c4b6814dd7f407ab75d1068cd295981 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/91ded3e535634b0994e97e023f79eb42 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/14af3c38fc334e77b2bdad7b3a003e74 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e60e4d84547c4e28ad30507c3142e220 : SUCCESS in 1h 18m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d406078dd6ef40c89117b4e1b5f27795 : FAILURE in 37m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0aa4d5329e4c4c6bbb41952952fa4924 : FAILURE in 46m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3be2c8230eb64684a06b337dd460e75e : FAILURE in 45m 34s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9b941bd03ec425c85837d7e733fdc95 : FAILURE in 34m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8525d42cf5b943b38a2c18eec4638d96 : FAILURE in 31m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bce48dbeb4c5451cbe650375dee0b592 : FAILURE in 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ca0dcb8b53c4d7d8a39ede5c1dbcb8f : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c69cfa187674148a3d51bf9b7dc9dd8 : FAILURE in 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f846f2d47c34b4490a25c903ff7f6e00e207a14","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-23 13:22:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b42864cd157c7fca9f053e0dc3bfe6664e67b38e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 15:32:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0beba0bc89a0494e90ca90bae194bd3a : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e24b91387c4405397f6b2e699c4fbc1 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca45374eba64ad29e43af98c1eb78de : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/140d9ca609a244068f9f00ee2822d426 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d281b2ad09447daa42a319d6ab24c89 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e866d63df44aacba32088b8efd03bc : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59668d27522845a38ccd152594895008 : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71450c2a0aa14ae681bdeb9d29cbc36f : SUCCESS in 6m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d413f04f617243a0bec02c509f66e9e4 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ed4d7ccb2e2416f993a7437ab5e8e29 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11f394992f4e4821beba8e3b53aa2fdb : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27890fb05b2f4aa2bfd40ae82b629d13 : SUCCESS in 1h 31m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f019e74cd4564b9896625e21071afef9 : SUCCESS in 1h 02m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e7a19c12454440c1b2e854911f357ea6 : SUCCESS in 55m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fb361526aff480ab9c9d54b35bbfb0a : SUCCESS in 55m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8c239ae1e9a4692821b0c11a4067376 : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80f69011716846cc96ed64d0ab423645 : SUCCESS in 2h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/89075055b80446cfab122eca7525c558 : FAILURE in 48m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e9e6e3c0fae34f7dbcd7cce549bb3458 : SUCCESS in 1h 55m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"521baa662e3f2ca12fd7afd5d7e5f332e58cd4e0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-24 09:27:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eca98c3f9a0976959ed4ea173b273a60f499ac42","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:10:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9b99eb6955fdce6834750460154c9a5e5885835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 18:10:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"381812a82d7531a8d1671416932823d94749cff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c03eb24b200d4c6b8b0cbc259896dcaf : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9441a723c04b0392685d31635e848c : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6705afd46f2443ee8974eac2cc7ba474 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0941569756e44a8992f7e5972afa0631 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a605a42946ad4b64b641c1a533978607 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dde1e4c86c64e849f068fbb2e402062 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee12511b0acb46d29e65ee3e909cde7b : SUCCESS in 7m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db50bf6714e54e719f9ef84ea8ec883b : SUCCESS in 1h 16m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/618b9c10c7294db7ad640a8aa9f952d6 : SUCCESS in 1h 43m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/23ae435a2f1941b5938b54a1099bb496 : SUCCESS in 40m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fac75d29da4c43f3a04889820d734425 : SUCCESS in 38m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be10d8281e3e457fa95b8b169f539020 : SUCCESS in 46m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68f12b3d842349f4bca249803ab61dda : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"037635b99b6d418f6e5452ad69ab5c697fa122d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:08:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4ee2f4c77bf2af6170a6735c30e1fab90bf67f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf7d5ff11be4393951c43a70b1b224e : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3301eeea01dc49d3940bd91d628d2c71 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f1307affc328e5ca58a2e814ed55885ed0710c7","revisions":{"9f1307affc328e5ca58a2e814ed55885ed0710c7":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:14:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/00/773600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/773600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/773600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/773600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/773600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/773600/1"}}},"commit":{"parents":[{"commit":"6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe","subject":"Merge \"Add missing log line for finishing amp operations\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-28 16:45:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:13:33.000000000","tz":60},"subject":"Fix nf_conntrack_buckets sysctl in Amphora","message":"Fix nf_conntrack_buckets sysctl in Amphora\n\nSetting nf_conntrack_buckets in the amphora namespace fails because this\nsysctl can only be set in the initial namespace (cf kernel doc at\nhttps://www.kernel.org/doc/Documentation/networking/nf_conntrack-sysctl.txt)\n\nThis commit allows to set nf_conntrack_buckets in the initial namespace,\nthe value is then inherited by other namespaces.\n\nConntrack is not enabled in the main namespace, the new default value\ndoesn\u0027t affect this namespace behavior.\n\nConflicts:\n\telements/haproxy-octavia/post-install.d/20-haproxy-tune-kernel\n\nStory: 2008028\nTask: 40682\n\nChange-Id: Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615\n(cherry picked from commit 64a301d4ec85a1bc9b208e60d7958a89e6972976)\n(cherry picked from commit dd1580d332c2191e8d28bab692c1fa08e3e8ae80)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f1307affc328e5ca58a2e814ed55885ed0710c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f1307affc328e5ca58a2e814ed55885ed0710c7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
