)]}'
{"id":"openstack%2Foctavia~773601","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615","subject":"Fix nf_conntrack_buckets sysctl in Amphora","status":"MERGED","created":"2021-02-02 08:14:43.000000000","updated":"2021-03-01 11:19:14.000000000","submitted":"2021-03-01 11:17:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773601","meta_rev_id":"fab47943c1e710e52ac1995486640bf0f2f2eb97","_number":773601,"virtual_id_number":773601,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 11:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-01 09:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-02-26 18:10:20.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-01 09:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 09:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 18:10:20.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-01 09:49:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"83d9fb0f54edf7206ea4253d86a1d0915a4b7c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:14:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"858961ee77e96676e372d8d16a295abfee28da01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 09:44:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aa94eddc04847ff86fb075e8f173230 : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8b55747ef5f410f857915ad045d6292 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abcefddd36646148af1c0777017a54d : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6109157fa3694242963671cffc31e965 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/295a8554f0534be6868f7a5c01a8ef95 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ea864baac684318a461e1195c2930a0 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4719f7c34ecd4d5ebd1ab8bf24c8b12d : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3223ee4fcad14889bcc883b88fb4ee19 : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/67ac5ca2dd5f455ea44bb37f29af9ea0 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/46bb6ea894854124bd7b7f8de5f8d65f : SUCCESS in 10m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fcffdc4954c74e45ba96b6ea286d9663 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04dd380061c34c81adcca0cdaaca5d72 : SUCCESS in 9m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c72eda2af444619a773c89963cb6495 : SUCCESS in 1h 21m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/94f8f270ff674efd841e038276409905 : SUCCESS in 1h 23m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17bdced185ce44cfb90d9aafd407a4f5 : FAILURE in 29m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef43af98e1ac4c46a3e2d0d1efccd0b5 : FAILURE in 22m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d7c912e0d126442b97e49d8bc2be7080 : FAILURE in 10m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b3ff5350f0843f3a6d6c86aa9cec7e1 : FAILURE in 28m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ba99ee9976c4915bef132d5bc51a62e : FAILURE in 27m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7492ab7b12664cefbcbd47f20bfe89c1 : FAILURE in 36m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/167c3b87f43f47de9a6d9585272ba7d3 : FAILURE in 26m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc1244f07a074dbc9ab2b5cc17b4e090 : FAILURE in 24m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9033abb978d417482a178be72c0b969 : FAILURE in 24m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/424ea45b93624df39caad60da78a8702 : FAILURE in 30m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5962b6ba3de74d98bfe684fd1bb5f3f6 : FAILURE in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1a2061b8e67a19675eebe1a3bd18164f4253e40","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 10:10:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9d81387b47b6daf2d76878408e5a1c1ad5ad891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:20:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c0b72b0150640478c59c3c5605d2242 : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd4c19739b4e4bb6bb029f04519a5adc : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6e7e53ae1a485c8dd582d974f2cf6c : SUCCESS in 9m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/301f1cf1de8c4f46b7c783f26ce79fbe : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6efed57ff5974bdaa70a87cabf273c82 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/745bae2bc5174d33b2a37a701b5e0c05 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1700313c324c7b95b3849b33282568 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23fc2550cfa7464eb2e5f5c4a2aeebfa : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4cd637c1ef8e45f2ad677b857318342d : SUCCESS in 8m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/83c734be59424490bae9f867fa317c6a : SUCCESS in 10m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/08d5694dcfcc4d6cb41a4c476d8a264d : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76f1d02925214eb0ad1f5811e83489a8 : SUCCESS in 10m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6575133163b34768abf26c381d3a0aa3 : SUCCESS in 1h 14m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a5182e80cd6f44cb86edc2e2743b6632 : SUCCESS in 1h 17m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6515e83587224213811e063409209bdd : SUCCESS in 2h 05m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c410c2688f5c4eb083b98d48aa99656d : SUCCESS in 2h 28m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7debde910a954feca2b86d920ad0fa53 : FAILURE in 12m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3918d4a22974b2f8f993b876d180bc2 : FAILURE in 1h 50m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3e2b57cba794ea7a73dbbe38d5a08f5 : FAILURE in 1h 27m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63b8473d9b7b4c78856b644156fc609f : SUCCESS in 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30add60a010f4327a4c2718b24a56579 : SUCCESS in 2h 02m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8873e6e3775d411688046ca0b7524937 : SUCCESS in 42m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6ec0511603cc41c6993cf022e5c57f4c : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a4da6cab3fae49fcacfd72362ea423c7 : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5b02eabce91c534fccfd6d23132b82c10500519","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:10:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea86bb48700154af54f5bd2e762d5b45c1b8f879","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:49:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2849792eea1b99e697f4f83b07b9d903b605d790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d410976d72b5a4fb5fb4361984d4934716d777a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:17:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78a85081fe3a431b8719daa1df1db2ae : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e414a904d44331b3bef4f54a036795 : SUCCESS in 9m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c404210eacd6462985c4fe798aabc61c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b50905508f4a87bccc7bcde391ff1e : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0086f60517b4698aa852f833537d8a5 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70733aa9a9749bab4ae39cfe34e743f : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87feeccecc254e0e8c0fd0a99eb6527e : SUCCESS in 5m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8339e047a85a412492ea1a9ef4b4bf36 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa64e115d1fa4e3bb7aa0a784e85dc3d : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5caf7ba6c4c04233adc6d791db5eea8c : SUCCESS in 1h 18m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a7a18dce075448e990c27f01c462b47d : SUCCESS in 1h 22m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db3369fc1d124e1dbf6a82486218bd64 : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e3e89fa564329b00220a90cbb0031382b23de4d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:17:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fab47943c1e710e52ac1995486640bf0f2f2eb97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 11:19:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7860f8353594d30aa04633472a441f5 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/725762e171fa4bd0a855082955f72e0a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd92b4bf414e0c7a34bcd295810b02128ab0d7c1","revisions":{"bd92b4bf414e0c7a34bcd295810b02128ab0d7c1":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:14:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/01/773601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/773601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/773601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/773601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/773601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/773601/1"}}},"commit":{"parents":[{"commit":"fd622009e6641e8b9d04fc66f1e878ef452d8d9f","subject":"Merge \"Fix amphora failover when VRRP port is missing\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd622009e6641e8b9d04fc66f1e878ef452d8d9f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-28 16:45:40.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:14:34.000000000","tz":60},"subject":"Fix nf_conntrack_buckets sysctl in Amphora","message":"Fix nf_conntrack_buckets sysctl in Amphora\n\nSetting nf_conntrack_buckets in the amphora namespace fails because this\nsysctl can only be set in the initial namespace (cf kernel doc at\nhttps://www.kernel.org/doc/Documentation/networking/nf_conntrack-sysctl.txt)\n\nThis commit allows to set nf_conntrack_buckets in the initial namespace,\nthe value is then inherited by other namespaces.\n\nConntrack is not enabled in the main namespace, the new default value\ndoesn\u0027t affect this namespace behavior.\n\nConflicts:\n\telements/haproxy-octavia/post-install.d/20-haproxy-tune-kernel\n\nStory: 2008028\nTask: 40682\n\nChange-Id: Ie6ccc4bf0017587df8e8e29d8ee3bf5c19e6d615\n(cherry picked from commit 64a301d4ec85a1bc9b208e60d7958a89e6972976)\n(cherry picked from commit dd1580d332c2191e8d28bab692c1fa08e3e8ae80)\n(cherry picked from commit 9f1307affc328e5ca58a2e814ed55885ed0710c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd92b4bf414e0c7a34bcd295810b02128ab0d7c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd92b4bf414e0c7a34bcd295810b02128ab0d7c1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
