)]}'
{"id":"openstack%2Foctavia~773608","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","subject":"Support deletion of all tags","status":"MERGED","created":"2021-02-02 08:41:06.000000000","updated":"2021-03-05 08:43:38.000000000","submitted":"2021-03-05 08:41:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773608","meta_rev_id":"a510124c9e60657aa0f85ea8a481ff70a7bdd12b","_number":773608,"virtual_id_number":773608,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 08:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-01 09:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-02-23 13:38:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":2,"date":"2021-02-26 18:09:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-01 09:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 08:41:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"state":"REVIEWER"},{"updated":"2021-02-02 11:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 08:10:36.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"CC"},{"updated":"2021-02-26 18:09:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-01 09:51:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-05 04:15:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"b7edeb73532beaa43e4c89a94559f89caf8b86e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:41:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdb83910db89b3af313f2fe47d3236d32184cefc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 11:25:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce2f541f1ad4269857bbe6abbf8b1db : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8c167c9c4374e57a322f2f87952e508 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b19f82de904df1a27b29cc65a11f9f : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/519d2409bfab4394a9360ada832fa8bc : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc265aad632452aa35fbaa7288a9874 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5aef3e1f854038b6f0f75321047ef8 : SUCCESS in 13m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea532dabe2c64947beae1c524f0911af : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dedb91db9f10428f83f5183b878c56c6 : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2299e9048f41414db3b24ce78f527aa7 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec5ed15499e24bbebaafc09563a3f427 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bc3c8d8bfcce4d8e8cd1b016843b10f3 : SUCCESS in 1h 17m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ecc399732064edd87106e4d30d3bd1c : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bec9b56fd6c6486b9ec5124982bbd26d : SUCCESS in 38m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ec6eaef85a44c39abeb7c8ded737a74 : SUCCESS in 38m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d8e9aa94fb46e483257781be89c5fc : SUCCESS in 39m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9e839e5f79c48ccb2fec1d297a3cf9e : FAILURE in 31m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9c5e9cbbc664169818cc98ed911495a : SUCCESS in 1h 50m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/187197ccd3be47d0acb9ddeab2448857 : FAILURE in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bca83204b956415f9b3aa72976dff2e7 : FAILURE in 1h 49m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ea787af4c014867be07dfaa3f187887 : SUCCESS in 1h 52m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d838423ab8024ec7aae0853a45099c72 : SUCCESS in 2h 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f28b16a217524574b2ef47200856333a : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb7275fd62f7388b149843e2b7f94e0e2a4a43df","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:10:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6c7eeb1b2477e9a0103d9f2ec64aa5c63b8f199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:26:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca52a6c4e7454c48b9483ff2d7b578ce : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8d9b8abd20d443da78eb4926ec9f891 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858d14a8d14d45bbb688f53482ab6a47 : FAILURE in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/587a1a82c7de496c8ff9ef67756db714 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42b603ea84cf4f4784c3b2675619513b : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ffe41a3e414141bf5cb5ff316eb47e : SUCCESS in 12m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/30947a9465f74b56b6d27d129b38ba1e : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08c396ed758746209da2ebbdf1b800a3 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/688ee34e0582491f977e1a150f900e5f : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a8f153701b8e4219802b6decf8c609b7 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2bb2f22d8fd4ff6a3eda0d455f277db : SUCCESS in 1h 10m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80639567804b4b63ba92232522c706f4 : SUCCESS in 2h 00m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41ce9c2a61474b9086a67cec7a27748d : SUCCESS in 40m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b52513c2c1145418b9a85eeef60a3d5 : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/656f5443ff4243198650a6ae83719c54 : SUCCESS in 38m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37758c2b93ef467c8fafa90c1b92d360 : FAILURE in 32m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cfec947e14a749f3a2cb7955abab5f4b : SUCCESS in 1h 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b846bce1c0b04d98987b68d5b4bfda0a : FAILURE in 2h 13m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/402cd56cda7d4917b32441b682975373 : SUCCESS in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f274d7edc8c340bd9aa9b73e44d1e8b7 : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/09611e79d9994bdca93e11854829d0b5 : SUCCESS in 1h 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3da68a041cb2408d9b9b40528c072577 : SUCCESS in 2h 05m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/policy.py","accounts_in_message":[],"_revision_number":1},{"id":"e8e820e32c633443be3c1eb3dd5040237badfaf0","author":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"date":"2021-02-23 13:38:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"411783e3b438c849f45b8eb88df0944df687a189","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:09:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6d8d4cfbbffb4f546d0e91fc317e4144847727c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"19bcaaf71261ad399a4220d5d501dbbdb0b4e959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:41:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/075855d084094519bc52f138043e5bdf : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c10da2ec3a8e42c8ad2472c4a5abbddc : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d694f88dde843e89f01e0f2cce87027 : FAILURE in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5230270253b743bc93d3009dcdf2d629 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df8680ac760f4dc39c50e27b768102dd : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720e37ab1aa54a5b897ab7fc93d0cd10 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/82d31b83b0ce4d1cb94e85401499905b : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8dc03639b634c2ba3c9cbd0976435b9 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7994b6d28bf546d1936b6d545d05ebcf : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1f86e5cc26a46d788dcda56f3a89176 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7a220fe1b9354ec195741e6b92fc09b5 : SUCCESS in 1h 23m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52ab824b28e641a6a58f9c5e21b2a8dc : SUCCESS in 1h 50m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac92086bcda44064a0c9a963f47ac8c4 : SUCCESS in 41m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b4f1f3977214e2f9f4b0dd26c499e46 : POST_FAILURE in 56m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b015cfa4b500497b81365e719aa47100 : SUCCESS in 38m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f03098f415cb4cfc83581df7f359210a : FAILURE in 46m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b6ba1ded624b4e71a8fce915d6092079 : SUCCESS in 2h 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d821e120a7c4786ad3b4af18f02edd9 : FAILURE in 2h 28m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/78f667a06f6c4562806740b245038aad : SUCCESS in 1h 47m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a8df86fc72942f7a6bb815cfa88f6d4 : SUCCESS in 1h 59m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b641b2aa8d6245569e8fecfce771cd04 : SUCCESS in 1h 59m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7c9ca1d8b1544793a50c812f9d8c56b3 : SUCCESS in 2h 12m 04s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"5c6f733cff33014f7b7dcb42fbcc596046f9807d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 04:15:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba6ca186bcb95de373638a40df3a577284f5e393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:35:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34063e1bf33c4821956b1ddfab72b248 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dc796ec44aa4647b3046fa54327c381 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f02ce5049924dc7b2f08f1249ccd624 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc3fe2af16614ddea6ffd2b7870ad24f : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb843e615134a238fd5f3853a43c8a9 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ee14f30ea445ac891043d3848ab47c : SUCCESS in 12m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e01f51c610a64b569a6fe1e871b2890f : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d96de693a9114ece91006908335c035a : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85e04d8636034d6a927e966af8747107 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2b0b799bce24828af8d4d9db48cab03 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1b378f2473940549789596c0c877384 : SUCCESS in 1h 16m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb885a30e4da422c861ff905f1a4a8b8 : SUCCESS in 2h 01m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b61ad29fb0e4fa2ae7716c573e6f89b : SUCCESS in 33m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eb72eb16926a46a5a8d265c5854b9de3 : SUCCESS in 42m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fc92b0d903949b489e68ab5cd8e69d0 : SUCCESS in 32m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74df6282410c4f30ae71fb169e902fdf : SUCCESS in 50m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbf5a2433d62497b8ad815a1bea9eb33 : SUCCESS in 1h 29m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/038ef14f429641ffa60215a5030716a6 : FAILURE in 2h 15m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d708c5a7cf90435789494c0ac0c8dae6 : SUCCESS in 1h 54m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cd901d76cbf3482da18eee79234606be : SUCCESS in 1h 37m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8dcc537bab3748e0a1ae5bd1b62996df : SUCCESS in 1h 48m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cb366e8e0e6f4270b6482a0ac671b73d : FAILURE in 2h 06m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71435dbe6a26d6e64e6edb167e86cdad32635823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 06:35:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24fb7c11cba0e028e8361d2380c66a5ff820096f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:41:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9de41a36e1de4a448629236ead6adde4 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb0bffd079d49b694c9fb2c1fc34bd1 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cca24be025f4f289da13820ec1fb88c : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ba24ce7c2d54f4b9c394e179e8764b1 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b972efcd08fe44258436bfc21bc6b6cc : SUCCESS in 13m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/664e6bbb4d414ae5914aa9192fa5f344 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a21d5329193428aaad93480257af97b : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b490a1d02a8f409db2306a638a9b2374 : SUCCESS in 1h 15m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd8d4bbc16054965b13e310819fd335c : SUCCESS in 2h 01m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f8e42b7a53347138873438877a4896d : SUCCESS in 43m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2995b8032fb451b8f0ace7111f82989 : SUCCESS in 45m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d79137cae92b48cbaf4fe5fd5c8d218a : SUCCESS in 34m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b0c13c478e542069b41a73530778983 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"56fb53bdd07ba5cca4878369c43db94d286445c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:41:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a510124c9e60657aa0f85ea8a481ff70a7bdd12b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 08:43:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152a5403f15a4b4b93eced5d75dff4a7 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adb4e78e4439181f57c17f277bad0cd325f219cf","revisions":{"adb4e78e4439181f57c17f277bad0cd325f219cf":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:41:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/08/773608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/773608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/773608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/773608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/773608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/773608/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:30:33.000000000","tz":60},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nAlso moved assignment after initial session query for the\nobject in the listener update path.\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n(cherry picked from commit 7ad022379ff2e366d3dfd73113d4619b0137551c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adb4e78e4439181f57c17f277bad0cd325f219cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/adb4e78e4439181f57c17f277bad0cd325f219cf"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
