)]}'
{"id":"openstack%2Foctavia~773609","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","subject":"Support deletion of all tags","status":"MERGED","created":"2021-02-02 08:41:57.000000000","updated":"2021-03-05 21:21:13.000000000","submitted":"2021-03-05 21:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773609","meta_rev_id":"17803995b10c25c8242efa22cb75ae7490c60e6b","_number":773609,"virtual_id_number":773609,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 21:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-24 09:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-02-23 13:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":2,"date":"2021-02-26 18:08:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-03-04 00:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":1,"date":"2021-02-26 18:08:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 08:41:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"state":"REVIEWER"},{"updated":"2021-02-02 10:31:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 09:27:50.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-26 18:08:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-04 00:32:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"25a82794a6ae57a27e6bc5ac326aea17ee1b7dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:41:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34d9cbb63e57593562281ef10fcfe2364571c42a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 10:31:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/add20ed0645649bfa7fb7a0ff5e62e5c : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44fd49c0c1254926934630cb666649ef : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aeefcfc89d94e80b855e0c8a1054ca5 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/978d86f41db34dacba85fb904d2185ee : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/077cc9fb9ba24311b44e85a1be5753e8 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652db0485b754d398d648d2bf3c4227f : SUCCESS in 12m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f6ff0ff0f5d542d396def224c45eda2a : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6bfb9506c57847ebba5b7ebcc5dd0f8d : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/717f175bea17485fae0be2b4e5693953 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/777cbae602bc41fa9354d9d95faf8b8c : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d17d8e9f4874b0288b4ce3215ad910b : FAILURE in 1h 11m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd98936e9ffa464c84cc8afe45ec9163 : FAILURE in 33m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5fa29193b51a49eb9a811558fc72683e : FAILURE in 45m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/333e097970a34f3ea32f4a4e07a73e56 : FAILURE in 29m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2439ee976fe243a2b0311351574cccac : FAILURE in 30m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/feb86111a89a47ffa5fbb1f0e4e0ca2d : FAILURE in 38m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b8886c9bfb4945558d14345755a773ce : FAILURE in 29m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/40bc9852a3ae48c59964d1f3e24d454c : FAILURE in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"831991c09d3910a5c83211a20514d6ee12274115","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-23 13:22:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc5f24d5acdb8ca651dd849c6d1a5b901c993edb","author":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"date":"2021-02-23 13:38:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"25400738b0b92571b5872dc834cb4976cfe3af8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 15:52:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/153be120c4264a7e84c5ee103c06883b : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e06e36f715c04d82bf661dfc37c0c5fe : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c902f2b0914d27ae429c9fddca9a5c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20475d47261f42b09cda9533b4074e09 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/822756b2e4624459a402baf023e375e5 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f459231fe5446eda5a4139d9769a216 : SUCCESS in 13m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ac0e5edf66574e4fae8ce0c0e2ec90b4 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70ca5988099c457a8710c6b5d8fc4557 : SUCCESS in 8m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22d0d67d4c44449a941ec535ba2ed30c : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b8e46e3ab774991b98de5170cf7dfd8 : SUCCESS in 1h 24m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1179097f97849069a81a9145b807bf8 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b68515b7f7044083b32658e430b561fe : SUCCESS in 49m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b20881c8c652451f9daa94fde356b8a1 : SUCCESS in 41m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab9248e4749e4a66aeff7df3fa0411ad : SUCCESS in 43m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7317440ed5bb45afa215bfb518106a46 : SUCCESS in 1h 20m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c4e09f3e1bb142d68de397eacab1df69 : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/808f4e042347446dadb0a56cf695644d : FAILURE in 1h 07m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5c4a73c5a03454fb21798389a4ae844 : SUCCESS in 2h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3e939505a116a79131c0558e2efe991d0557a60","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-24 09:27:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a55b4d30b74a6d8327b2a14301d5651f1d663f6c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:08:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27c222346c63a114829ed45f427b4112c4bbbecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 18:09:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9946b7371dcd0d9b3075fb9f18321b55ed4bd41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 20:24:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/494c278d49194444a2d282c3d0aa7b68 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053992ffe79049c99597648b2e3cfad7 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/797bfca32ca649059497e18ad18bf58e : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e1541fbaf7e4b8db4b29ed794e9a189 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90769b7e908b4af09b469af73997fd92 : SUCCESS in 12m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7860098c5b004f7fb9895e42885c96a2 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd82cadb28a84509b498c137a2407dfa : SUCCESS in 1h 39m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ab64486d5b2447da5657ff03b1c3f6c : POST_FAILURE in 2h 13m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/720b57a3d5e248cd8941fde31c1190ae : SUCCESS in 43m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9c899c7515434e29bc8389100bac5a97 : SUCCESS in 56m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e5f089b214c4fe0bb57052b4441d4f2 : SUCCESS in 55m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ce888b7a21d47aa9f8c1fd8f248ec94 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"252f24fff3eeeb3caa68942f79e5a60de8c0c3d0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:32:37.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71a56e622a97d72a38f6a7748b0531ae0ce0ef88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81c1a6a70cc54ad3a0fdfbf9fb2b9f19 : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dd8c7ff20374c0db81024454c411002 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49388fd10a4b4cbba11dd6eb56c1e049 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/318306cadf93402dbbeda2dc8df22128 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98a6f619b7d24757ab4271c86f34111d : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8ac4ef9d9244aeacc262623a0ea6b0 : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/88812f74c4194766a6799ada8d5c0f21 : SUCCESS in 8m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51f07e7fe70f4cbca41f0234f5bece71 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/680f1ec0b4ec47ad86ca97ef773d6fae : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cf67fa4b6f64195b19a1861873e77ee : SUCCESS in 1h 31m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70ea47a3437640c3bf28e80e1fdb6d3c : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80ee1a92ec1f461ba5cd3c2ed42bf127 : SUCCESS in 39m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7ab877569bac4ea6a2386b1d6026c898 : SUCCESS in 54m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3399137a4dd2429faed74e0c3fa34d22 : SUCCESS in 46m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33372dc040144c6ea086338db99ae413 : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07725b8f7b6c497c84c3cf6dfc551db1 : SUCCESS in 1h 55m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30b38fddd4ed4c91a802a9f99c6b7053 : FAILURE in 45m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e1397297b1a4959a3a64098d4ddf3c7 : SUCCESS in 2h 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8de7892f794666e8c3f0714f587c7071105ebca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:58:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a386ed6a44b7fed3c23ace0709ddd4fa392f016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:21:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/018e727132da46b7ba7e7cf4c312d7c8 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ded3408a92948279e20bbe92d674126 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ca438e996ec481c9bad9abc5a01404a : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/994960f21248455e861a8f8aaa0280c5 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d184396f4c4d2bbd00f898c7956047 : SUCCESS in 20m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7bd4a9608d2b4607871a54d9e75c65a9 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3681d299cbe64ab3b271c67706c8613c : SUCCESS in 1h 06m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7eacafb85bce4fc6868df6543763dbe4 : FAILURE in 2h 21m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f93d1d7ba2d749d58dedfe8e5f3a8b42 : SUCCESS in 37m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc6726e281cd4c49a7a3e0a141294645 : SUCCESS in 41m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/031e4ecea519452682107288c0eeaf5b : SUCCESS in 43m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68243457d94a48ee89b126c233eafa76 : SUCCESS in 1h 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ab601dee31d6a16247f7352e42ee0f16cbc22f25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 16:29:08.000000000","message":"Patch Set 1:\n\nrecheck\nnova failure","accounts_in_message":[],"_revision_number":1},{"id":"a99e21c2daf616d01e96b990c8b2563939351ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 18:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d23e345e1714402aca8eedaab6abca9 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a9d06d98094408192882561f98517da : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a485e2af307442ba1305eae5ffcd33b : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9871795c4fbb4991970917703cfaa7f7 : SUCCESS in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12a4d74e4e6c473983f77ea6c331d49f : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb644d91e474f28be51175108543e06 : SUCCESS in 14m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/041e2e66629d4d3c856e435d081f08ac : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/64e3ad1326b54659ada09c2bb5d320c0 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85e5ced2f86c4850891ecea7d79150ee : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d321b6aee9f24d8282b234e58a6dbb49 : SUCCESS in 1h 36m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/867193741e9a44df809f4dae37b93a57 : SUCCESS in 1h 43m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e8a2434e7db458294559cc1f33d00b2 : SUCCESS in 47m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e3a2223d65dd4d6dbc77781f026250ae : SUCCESS in 45m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dab9d0f9527b44fa8eb7b1e7a8ee4806 : SUCCESS in 46m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd91f8365fc5439b8cb3d87a942ea9b9 : SUCCESS in 1h 27m 39s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40a04013eabd4ed2b358412f1f31c0f0 : SUCCESS in 1h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e57af11dc0648aaaddf41a06174e70f : FAILURE in 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/82121835d232492e92ccc2aa2e45a79b : FAILURE in 1h 49m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f353dbd9d3fe36f8bcf0f6691c2d43b32ec7670","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 18:58:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a349f12d3784d81bffeac9ee9c11b92f9322821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:19:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dd34e81d0074bea81f66f6eb2b22e7b : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4e0a1af0624c519b0e89f6341725b4 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27394deee374232af7a4191bfb6199c : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5b96ee447c743988dac46e46fef62a2 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4cd05312e44a34bf83093b1cdad700 : SUCCESS in 14m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15e80054391f498ba27da184b63fbb22 : SUCCESS in 13m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/36d25807a1b54ff493a1800569f7b32c : SUCCESS in 1h 47m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ed034255a734e1e98cfb52d239d7ebc : SUCCESS in 2h 17m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd20c7bd0c9c49ea9ea04d1a7e372089 : SUCCESS in 50m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6443ead7a193459d9c0feae2abb7c42d : SUCCESS in 41m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3622815edcf423bbeef7eae81ced8c7 : SUCCESS in 46m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/188b16937e634e069b107e686bc5af46 : SUCCESS in 1h 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cfd5065e064e8ba8c533b813ccb54c90dc9527be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:19:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17803995b10c25c8242efa22cb75ae7490c60e6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:21:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1204943fc642d5a021d0b4b53b23bd : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95a97cf1eba85222e4ff683b88878f0f030a2f11","revisions":{"95a97cf1eba85222e4ff683b88878f0f030a2f11":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:41:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/09/773609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/09/773609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/773609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/773609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/09/773609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/09/773609/1"}}},"commit":{"parents":[{"commit":"6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe","subject":"Merge \"Add missing log line for finishing amp operations\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:41:39.000000000","tz":60},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nAlso moved assignment after initial session query for the\nobject in the listener update path.\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n(cherry picked from commit 7ad022379ff2e366d3dfd73113d4619b0137551c)\n(cherry picked from commit adb4e78e4439181f57c17f277bad0cd325f219cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a97cf1eba85222e4ff683b88878f0f030a2f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95a97cf1eba85222e4ff683b88878f0f030a2f11"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
