)]}'
{"id":"openstack%2Foctavia~773610","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","subject":"Support deletion of all tags","status":"MERGED","created":"2021-02-02 08:45:38.000000000","updated":"2021-03-01 12:03:10.000000000","submitted":"2021-03-01 12:01:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773610","meta_rev_id":"e1e7eda6ab2f83dfd240a4a5d394b2369f097722","_number":773610,"virtual_id_number":773610,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 12:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-01 09:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-02-22 13:15:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":2,"date":"2021-02-26 18:08:09.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-01 09:51:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 08:45:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"state":"REVIEWER"},{"updated":"2021-02-02 10:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-26 18:08:09.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-01 09:51:20.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"82ebc1a31921299cb942a506e06458c6e7305c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 08:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba2ee87129e824a2218cb3b16674e0d96075c8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 10:36:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0c4c96558d4bb6a6afc89bcec3a8fd : SUCCESS in 10m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea43dd225e0949f1a1d353d022d30905 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ffc84381b324223a9dea80c516fa006 : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/005b6886864f496f85f2dc65aab9c8c3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dee7962a312498b9a79268b2b64f840 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbce6331274a4e9d836f07147c319805 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac9461e13b749d2a5d38e0a5704cd33 : SUCCESS in 12m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9c3508636de64fc3b5ee056108be70d6 : SUCCESS in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/75e54226e9e74a92af882190c30d0213 : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c67305ac98b34ab7ba666c5467d7b53b : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3ddaa41761248469ac01e2fb4cf016f : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5786064b85e540cdb1d9a7293d234c77 : SUCCESS in 1h 13m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1b4c6c67fb1e4e1aaf5f1d30dfe5d3c0 : SUCCESS in 1h 30m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79e749b3a5b54558956f65ceaed108dc : FAILURE in 22m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3cd7ce6123c44030bf3ebae7f15984d8 : FAILURE in 57m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/93e1a1656c614025a12fd77f987a3feb : FAILURE in 10m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec532454fd8c4d279e192048a37b4b2a : FAILURE in 25m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d6b9ea5f8f844d180a98478b6ecb300 : FAILURE in 30m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cefcc1bb40b64db39a4a178713ac6085 : FAILURE in 35m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17ffd7d42dc84887a08bd61343dfc12e : FAILURE in 31m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d83581d0d7e84e92a451721be2e75331 : FAILURE in 24m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/722fd06baabd40c9b047be8a317e7abc : FAILURE in 23m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d514f2177fdf4636a5c07565591b421a : FAILURE in 23m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1cdb8eca8a16469fa2fa2a7b9b838b6a : FAILURE in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a6f7e03bc43686afcc69b56d12093b67f0acd45","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 10:10:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f5ed34bce23e4e2a66b1598d608eb28261bb31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:00:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5b02659df144024af06da2bcb591316 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c530a03114e46599d64a8578ec6268d : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5810670f8854b588d50517f69034e78 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d642a4ee4a7c4823b5956e67c1adda6d : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b62b3651bc41afa24015794a1f826c : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/894979885a3447638ef3fd9a3a85df89 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d042e7df87e14cdfb43a5a53cae1b5c2 : SUCCESS in 10m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e086a04390245b4837f3a7917ffd7f6 : SUCCESS in 8m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5ad81d6b84864c93938ff1148504da3d : SUCCESS in 11m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4574cd837dad4b1ca81c0838088b1815 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c826233182ee4c3586931144c67dd350 : SUCCESS in 9m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af94db5048c84c4291afd57910d316b2 : SUCCESS in 1h 13m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b89cde1533584e74bf824e74ba3cb214 : SUCCESS in 1h 18m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f1cf629e3244e9f910f783eb3423c23 : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8f1f5a559240451da0a34705b642d0a3 : SUCCESS in 2h 15m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cfdd56e73de64793ae6fc2297899b0a1 : FAILURE in 13m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d3738f8bef26475ba57a4e38f41101a3 : FAILURE in 48m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd0243240b164cb3a710ef110bddf13f : FAILURE in 43m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b6b5e9d395e43219dae99ad432fb8c6 : SUCCESS in 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a4e1aa5ba3f445f3b2519bc1e4a2aeef : SUCCESS in 1h 50m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e389ea2684474c8581fec0bee8648823 : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e17c32bd01f7491a8f938679b055e248 : SUCCESS in 43m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a81da696cf9f4cfaab1fe2cba96db87c : FAILURE in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a8f7025618fb5c37f1459859c58d4ed340fb0a7","author":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"date":"2021-02-22 13:15:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0589c17ae6635033c145ea82a051ea3e9468a9d6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:08:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f232386c56c2fecc99f51b45f2077caec8d53ae1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:51:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d18bc818674f1e96a86f6b665c1d101d47656dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:51:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ede001380ec674c321340c8bdf3635a6c309542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:01:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0958e654310f48b396cc60dd426af03f : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28713670090f4979ad8e85162b6feca4 : SUCCESS in 9m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46f4cd1e526e458988c5d789160df6ed : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbec57ddf3f14f06862b20a40dfd5cfd : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ee258024bca4443aa11ad21065e35d7 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144951d6600c4b0ab8121f2917b54948 : SUCCESS in 12m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5922e089f57a4d74a77d507ec25d1577 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75786c3f5f2844c1b47b08dddeb329eb : SUCCESS in 10m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d6c31f3d97540919f40d421983fd08d : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9128c492697443a9b04d46f7f231c2d6 : SUCCESS in 1h 20m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50998cb2dd6a45ddbdc043a8b0302ddb : SUCCESS in 2h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b0f7244797984aaff2ecc8a589a9a1277d682d87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:01:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e1e7eda6ab2f83dfd240a4a5d394b2369f097722","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:03:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b72be561aa4c879e8eae81199ecef3 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a94e5af899936e3fb7e554cfccb4d66903f34c6b","revisions":{"a94e5af899936e3fb7e554cfccb4d66903f34c6b":{"kind":"REWORK","_number":1,"created":"2021-02-02 08:45:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/10/773610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/10/773610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/773610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/773610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/10/773610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/10/773610/1"}}},"commit":{"parents":[{"commit":"fd622009e6641e8b9d04fc66f1e878ef452d8d9f","subject":"Merge \"Fix amphora failover when VRRP port is missing\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd622009e6641e8b9d04fc66f1e878ef452d8d9f"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:44:42.000000000","tz":60},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nAlso moved assignment after initial session query for the\nobject in the listener update path.\n\nConflicts:\n\toctavia/db/repositories.py\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n(cherry picked from commit 7ad022379ff2e366d3dfd73113d4619b0137551c)\n(cherry picked from commit adb4e78e4439181f57c17f277bad0cd325f219cf)\n(cherry picked from commit 95a97cf1eba85222e4ff683b88878f0f030a2f11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a94e5af899936e3fb7e554cfccb4d66903f34c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a94e5af899936e3fb7e554cfccb4d66903f34c6b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
