)]}'
{"id":"openstack%2Foctavia~773615","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~If403a90ec27721ef63b6ae666fd77dfc9df0295e","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"If403a90ec27721ef63b6ae666fd77dfc9df0295e","subject":"Fix PROXYV2 pools","status":"MERGED","created":"2021-02-02 09:07:28.000000000","updated":"2021-03-05 21:26:33.000000000","submitted":"2021-03-05 21:24:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773615","meta_rev_id":"9eb9e6ef2f83feabc98b1524f9cf1ee5316efc8c","_number":773615,"virtual_id_number":773615,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 21:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-03 12:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-1,"date":"2021-03-05 11:07:41.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-02-26 18:12:56.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-02-19 18:20:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-02-26 18:12:56.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2021-03-01 09:47:49.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 09:07:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-02 12:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 12:38:33.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-19 18:20:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-22 08:14:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"CC"},{"updated":"2021-02-26 18:12:56.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-03-05 11:07:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"91b6ae06dd2d9a7182594f86152cd8ca0022033d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 09:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c19725a0581ccbfa6100db26e4c67c17a42308b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 12:00:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f99d94a3b97470a90dca50dca367a11 : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d107d432d9c6434fa329f835abb8e3f7 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df282782de44ce3950ce0e985a48ea6 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbecc63e0b1f4bbba5059f6121018b18 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92d11015abb7431c88f251192af9ca85 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a33777e4cf4959ac7ac2665489990c : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ef84d04cda4f99a5dbe90dd8a07b4b : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/378dff6d72fe43f4aedac26d051654c9 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9466a084e8a14bc2a024d79d3731b385 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b67f1ad51a84cdba141cfc3d75660e2 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00b5a5cbf6c04310a1091d489679a694 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/831ec2a2bc4a4bf8ba5ddf66590f27e7 : SUCCESS in 1h 21m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b74aca9f40ff4d0b9c91164477904153 : SUCCESS in 1h 37m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c8f2f0f1396a43b9aff5aafe31ec19db : SUCCESS in 41m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f3a8a0fe8646426f9b07519f0bbe827b : SUCCESS in 38m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99429248add7494ebdcc7905b32bb453 : SUCCESS in 39m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24ba0174dc3545cb83d5104b450eeb76 : FAILURE in 31m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8124ec2f5efd429ea7c0792e08424738 : SUCCESS in 1h 38m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df769b03ba2d483ba37a57101fdfcbb3 : TIMED_OUT in 2h 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09f37789919e4797bb47cfdb08a18e58 : SUCCESS in 1h 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c379f26afba449fb2916cbfee1eec81 : SUCCESS in 2h 13m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/58c939c3f05741d68eeedb069245d09b : SUCCESS in 1h 32m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d7b7c258f8204bd09832137ae90984e9 : FAILURE in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d5e8334233e4a4b85cd8beba1306709aef21324","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-03 12:38:33.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab12762fe9b91d0c85c90299b0074e6a6ee79049","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-03 12:40:07.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9983146d7d6708cd68c490490f5e446f847567cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:35:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c00939108743b7b24c7bedf9a601a8 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3fee0539b474da6910fbc6816ae64e5 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1bbf893f4304bafa54755fe2d57ef08 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/260d1225d25c40cf926751b94fccb682 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/036658e5ae42447f8dbe65c2aa706df5 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a11034e8414480bbf6b40ccaa8d574 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/426b3c7ec98d49e79840135ee4ed2aad : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a97eaf6a05e647ef8eef535cbfe3f1e1 : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8e44ab84908d4fe990eca964e6e76ba9 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2efd7cb35bea4980ab7c845c80105c7b : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5455aee80a54486d93bf12eef7755a9d : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad6c23e47f734d128cc6b963c0cea3d4 : SUCCESS in 1h 13m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/661c7228c8334091b3569f7e1103eb49 : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/826207cb511942f09ccbcbdb2f0d1a09 : SUCCESS in 46m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2a559e475b654b9abdb09894339b4d5c : SUCCESS in 43m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1904679dd1ca4a91a25518a99d3e51fd : SUCCESS in 42m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b58b18644c9c453a9e15326059132931 : FAILURE in 34m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22a9392342454ace8fc9e2ccf319b3bf : SUCCESS in 1h 55m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b532ac227f44873ab2f099763a76f72 : FAILURE in 2h 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cfac084c2d1249b082810023c243b737 : SUCCESS in 1h 27m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d17686ce95684b44af80d9b888d3d749 : SUCCESS in 2h 15m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f26b72f9245e46f1a261d69fc1ce49f3 : SUCCESS in 2h 07m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a0a4bc8c80844f678d69f9ad1446a9cc : FAILURE in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7adf266760cf15e4e77b826caff981a11377bbe9","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-09 13:03:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37297082ac7bd581e2c20c2fa860831625a5f47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:01:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c6dde2dad924f64b62f7b7ba5d69668 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26694d068e1c470089730e4dea3092f8 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dcac8526daa430bba8350c386d7ee79 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ce606797974d5a9871619b75b5e4bc : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d4db9b6c594defa362f92fd1b5930a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4a378ff63240238c19c7c127c30aa4 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246e493ddd784631b91a93a1c4ef9453 : SUCCESS in 4m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f3e01b352a34210b908f177aa622c78 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a77bb6945c84fe8bfd68bc8532c71de : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1bc0b965f9bd439eaf3c580d2b27858e : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b30a2e106744da381b643c86dff799c : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d17f4233942242499adb9c26120401f4 : SUCCESS in 1h 34m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16cbe3a1c10347bf824472eb05cde3f3 : SUCCESS in 1h 49m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6749ebe4e7b43c984e89c00734018ed : SUCCESS in 45m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5106648fb30b45cbb680e2172da1e09d : SUCCESS in 42m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/600926383a8c4673ac364ac90ec84fbb : SUCCESS in 42m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a91fe048b13f4ab49eeecb913e726a43 : FAILURE in 34m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a57ceea63d444a738cedd8fca8aeb562 : SUCCESS in 1h 53m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2182ddc45db54052a1035be87e420a2a : FAILURE in 2h 18m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0a670a4a3f2247589fa2efe20ee52c80 : SUCCESS in 1h 37m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a3a24da9844f44588fdff9dd51f413e2 : SUCCESS in 1h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2e6f59592382437297eb74a4604d4194 : TIMED_OUT in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0517fc1096dc4e8595e33795fe003cb2 : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2b433c6dbf01e1fadfd7e75f19c0208704f701c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-19 18:20:27.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8977183d8130250f2684218cb451aeb04eb994db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:54:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2481434061d490bb28ed8b09da6c9d3 : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b959b756ee244088d1be9e0463eab7b : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541b97a80da74892b198690328754156 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1086ce1076b84f9993d0ef8d026ea7b9 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84a683f0689646118c9f9d9b685d9d80 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d711a9ecf1408289c451c79ed0afd0 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f7d5697e6444d1982b62bc71bbade46 : SUCCESS in 4m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/054b4ef1a9334ab18517fd7e5b97c0f8 : SUCCESS in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c4d165a52d164aa699196c1fe36ab57b : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d11db4aa28c43e48270d1ed29045767 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f83c6d80d9954c5cb46ed89c9245185d : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65dec96912a7497593ad34882e5274d2 : SUCCESS in 1h 21m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c7c85e62a534fb8995c12d003459144 : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c9a0d84175a4c39a2e457d8ae263a72 : SUCCESS in 41m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/781c4cd613144743833710d5b40455a4 : SUCCESS in 45m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/899d9cc0b542432eacae2e71cdd6a33b : SUCCESS in 42m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/915edc5f6aa944068ded756f72a2eb35 : FAILURE in 36m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a6db2f8f51c4ac3a6e75a1d72abec60 : SUCCESS in 2h 20m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3332003a289444c4959e54f6edcc20fe : FAILURE in 1h 44m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9f7d90bf4c5542d4b683a9f056703f4e : SUCCESS in 1h 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/529a8b1c559f4e3c830f18dbed5dcb27 : SUCCESS in 2h 15m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ee575273784b4b74a50174febc82f2e7 : SUCCESS in 1h 52m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cff8cc234b284b4185937be9e3a6616b : SUCCESS in 2h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7dc94c2a5e21172d8128e5ac8a8e4babe68f0b9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:14:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd5d89552db360ae71463c9210d1f5da0948d5d9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:56:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d28dd1a728c4446d02767e6bd98365edd7b34192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:47:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/556128192a7f459bbe3fede28ad0bded : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/580269f991f64653ab2825b507b4d6d0 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c17ab8abf0422e811026b9e2f87f56 : FAILURE in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b238db50c43a4b3c99b5563f40319d2f : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98b1dd2070944e3827693b2dc46050a : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43042112d0f64960b223e3074aec2fb7 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b15d6b4b5b9845399d864e8a4460580b : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/684fe98669f3455e8d17bf7d2cb2f2fb : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ec3ad1f87144c959bd26b159314ae89 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46002df3425c47afa88990f1c6e8a85e : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4109caa83b424cf4a63bd9df482bfb54 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c174df931072453c9958cfe5b05fc413 : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34d26b4448b34462b07fa2a95d13f951 : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5318c0819de4c758739e5252fe13f03 : SUCCESS in 48m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b68ff0faf18a4c18af88c1f40d8da8b6 : SUCCESS in 40m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a493e3acc114d068793f2d6358686ef : SUCCESS in 40m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b551cf999c934ce597640d44feddee98 : FAILURE in 44m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d887942a56c24e3dbd4a370e3704f2da : SUCCESS in 2h 00m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23e6e940b24644c4be57193ff076a300 : FAILURE in 2h 22m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/abb30095377a4d3fb60bbc025bb98a50 : SUCCESS in 2h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a17f95e004d4489a38d89412c3b966d : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ab66e835fe5c498f9abd43770ba737ed : SUCCESS in 2h 04m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/55487adea52f4302bd6b1c151046126f : SUCCESS in 1h 31m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/policy.py","accounts_in_message":[],"_revision_number":1},{"id":"0290050f982754ec05f15d93b3ea117ccdb0a996","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-26 18:12:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nApproving, though this will likely need a rebase to pass pep8.","accounts_in_message":[],"_revision_number":1},{"id":"f180fa1098d168e83f6a25328d2594a78cdd50c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 20:36:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b60dd47e00c4ec4a0e8ba7621ca117a : SUCCESS in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a59d51980c439695aeeed3978c51df : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5849a59e943040afbd7517a0b24c25f8 : FAILURE in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a443d3d2d21f44a8acd9c22c801b672a : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d1bb047886c4b7293aae6712923177f : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0286427ed4471a8a6aadfcd95b380f : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2483ff8b3084dd0a9d931853cd308a9 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4168a450f0e84b7d8c02d82eda6efd84 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/06a2c9b7489f4564a931bd5c713a65b5 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4af82ae97bd9496db32b861c83484ca5 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b047f472dbf047e98f578e9c00464dd5 : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31048f7d675b4a2880122a5830f7e1e9 : SUCCESS in 1h 17m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11a7b4b978d348faa36bc99974119eed : SUCCESS in 1h 50m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e54b0118b6be499ea588568e7c3786bc : SUCCESS in 1h 00m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/935fd101daba468db28e94642039f321 : SUCCESS in 40m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/557196a9635c4bc7a234113eb0545714 : SUCCESS in 1h 01m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5de221645c834603b72d05682ac129af : FAILURE in 31m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f2894d90d714e99b527e6e4c931c0d4 : SUCCESS in 1h 51m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35829b7370b04ddcb3d7e638c3dba96e : FAILURE in 2h 18m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97816a73eb204dbe938e894efd00ac35 : SUCCESS in 1h 56m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aa8282b3590e4b17bd04ef3a2b1a0ea3 : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d03126887a534d26a8391a3093eea903 : SUCCESS in 2h 05m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fb7d4d5e2665417ebe64d79389dd17f2 : SUCCESS in 2h 20m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py","accounts_in_message":[],"_revision_number":1},{"id":"860b68b56655b7a98e3eebe8ba39ed4e086f6dcd","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:47:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"546efd6db301b25bfb34b04dba18421c13311d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:11:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ec8722720884413a5c9cc5147e7cd6e : SUCCESS in 10m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9c9c3da1c8245e59a6e8d0faa192ae0 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8a7cdbb17346798801a631363b4709 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2383df6e9cc34ec7b44b6be41f7cfefb : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94d8eba27c044ea29d343f2f81af375d : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4aac1e4b05f48659360f4e200435e36 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256f7a5c2ff84ee3b655a7b7653ef8cb : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07a8d76824ba44d4ba958275069f3a2a : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08108565b5734b7384a5ababb73b9572 : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d6ca937f34a4c5fb1368218660d2f1b : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26f9b57db42a41b98bbc2aa942e8137d : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5de5bef4102848b4a65ed0829fc72811 : SUCCESS in 1h 10m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3102d3ee1aee43888f7d29bc233347e6 : SUCCESS in 1h 26m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/97ddd16066f94d939635830c70bb25b7 : SUCCESS in 35m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4a092ab017674a768d642bfe5c647b2e : SUCCESS in 35m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/674a526098af4df88617102a33b02668 : SUCCESS in 37m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/181ace2a86f34785b9482bdd22c59f16 : FAILURE in 33m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/209b990331604086ae5ebff405ead78e : SUCCESS in 1h 37m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/28b6e1c5fd9149f6a91911ab59006b5e : FAILURE in 1h 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc74ae8eaa84bb09ba1e4f4b94960e6 : SUCCESS in 2h 02m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/52b02e21b0724973932716ba8da35f8d : SUCCESS in 2h 07m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/71c5f33dfbe643a385745bd5244e1954 : FAILURE in 2h 11m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cde12c90fb2e4422bf698a93126ae585 : SUCCESS in 1h 53m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"a0903c0dd84c12c7bdd86be2226bfa40ba07c368","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 11:07:41.000000000","message":"Patch Set 1: Code-Review-1\n\nWe don\u0027t need this fix in octavia/common/jinja/haproxy/split_listeners/jinja_cfg.py also?\nI honestly don\u0027t remember if we have essentially deprecated the split-listeners thing, if we have then maybe we should just remove it so we\u0027re not essentially leaving an extra template around to confuse people? (I think this may have originally been my doing, too, sorry about that -- also don\u0027t mean you need to do that in this patch, but maybe we should keep it up to date for now?)","accounts_in_message":[],"_revision_number":1},{"id":"6b4169a29c4fa513417af7fc0ec54f0261193a4e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 16:33:19.000000000","message":"Patch Set 1:\n\nAdam,\nWe decided to not add new features to the legacy templates.","accounts_in_message":[],"_revision_number":1},{"id":"156725cea5900d35395458963c1b4264d51d521d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 16:33:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"769353923a5d960acd46ec8d98e271e43a82f24c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2659de833d04d009d436d712e62dd1b : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6115948c43d346ce84a7c17b6770261e : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8487335a411478f80e3a86d756040c3 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f3bfc6c2ea466fbc2f6ec7489fbaa3 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac9b5ffdf1d46678488bbd4a0989d6c : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccd0261756934a6bb369f8ade7e8ed2b : SUCCESS in 6m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6dc870188772473a97bd4a7e5f0cdf6e : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bca7a571d6eb48f885c2cf732b628cc4 : SUCCESS in 9m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/873619f3e1464642b4751cf3288a7b27 : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eddd3c4823a2437f89f967c57ee91294 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fe89ec81eaf4734972a70b8a23830e8 : SUCCESS in 1h 35m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd709b52f5a14fd4b286ea1577c66a4d : SUCCESS in 1h 42m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1594f7d6af34af58c17eea506a85c56 : SUCCESS in 44m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95615927215440c2b13e185a67df5226 : SUCCESS in 38m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15228345987e44cab3ca98039c081ba4 : SUCCESS in 51m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad8233e8112f45b9ba22530bb41fafc4 : SUCCESS in 1h 11m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43b9c02f321c4c5686b526f747859e7c : SUCCESS in 2h 09m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffde380d399c4a8eb58c72c621c83ec7 : FAILURE in 2h 15m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cf7b4d566ef4a54b1362efa9055d515 : SUCCESS in 1h 52m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b72fe55b54344b08d2d1c37482f2c71 : FAILURE in 1h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b987e70309a54206825b14611778f41b : SUCCESS in 2h 06m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8e5a72353be6401aac699714775112f8 : SUCCESS in 2h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9025eec34632907ef9753f08c7ce37efe6af4ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:20:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91ff3b2dd25ec54698b455aba05da8d0ecf41906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:24:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33cb12c560e346708127e3ba79ae6c09 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d535566833d42eab82423709d981131 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccbed5af406d463ca2b48e8e09ecac9f : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3447e9ef4e5423ca11299aa4ad80f6d : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/494ae06b66cc47ae9db9e966c6a1592c : SUCCESS in 6m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c5a773fc67a44eb8f5ac53d6b3f37d4 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/828a8dc886ff471c92abaa7b07802f85 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/130a5a511eab496a98390a6351f949d0 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9c46d716aa54b3588279397730d45c1 : SUCCESS in 1h 59m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c6dc5dba65a400886b2837dc7c6c5e1 : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6b74cbf9c92c462ebfad81032f9fc574 : SUCCESS in 42m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29e55eec69bb4499a74881ec632e70fb : SUCCESS in 56m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/378a8c04afc6453da554ea508a3f48cc : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c77bfda4b2ceaba32f632b58b59468690067441e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:24:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9eb9e6ef2f83feabc98b1524f9cf1ee5316efc8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:26:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0019f97d00b24d5781071bfd04429e49 : SUCCESS in 1m 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a229ddd673a42559480fbd2e4f14883 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb54ec51eeacb31f31c356eea46d2cf0b56c5c25","revisions":{"fb54ec51eeacb31f31c356eea46d2cf0b56c5c25":{"kind":"REWORK","_number":1,"created":"2021-02-02 09:07:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/15/773615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/773615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/773615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/773615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/773615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/773615/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-09 00:10:35.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 08:58:25.000000000","tz":60},"subject":"Fix PROXYV2 pools","message":"Fix PROXYV2 pools\n\nThere was a bug in the initial PROXYV2 patch that causes pools to\ngo into ERROR when using PROXYV2 protocol.\nThis patch corrects that issue.\n\nChange-Id: If403a90ec27721ef63b6ae666fd77dfc9df0295e\nStory: 2008237\nTask: 41080\n(cherry picked from commit 4024536621616c6df88686d80694b48dec612010)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb54ec51eeacb31f31c356eea46d2cf0b56c5c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb54ec51eeacb31f31c356eea46d2cf0b56c5c25"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
