)]}'
{"id":"openstack%2Foctavia~773798","triplet_id":"openstack%2Foctavia~master~I602418264e171a2b1a926eff0b1f9e6dc186295a","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I602418264e171a2b1a926eff0b1f9e6dc186295a","subject":"Validate user access to vip_subnet_id when creating a LB","status":"MERGED","created":"2021-02-02 20:26:12.000000000","updated":"2021-03-27 11:45:38.000000000","submitted":"2021-03-27 11:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773798","meta_rev_id":"204aaab9b89e149f1ddf7b30963be4d2867f73f3","_number":773798,"virtual_id_number":773798,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-27 11:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-27 11:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-27 11:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-27 11:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2021-03-27 11:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-20 20:34:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-02 22:32:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 13:50:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-02-11 13:04:44.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-17 06:50:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-18 14:51:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-08 07:50:04.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2021-03-20 20:34:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a9f5bf85e1f6328805b5ef8adec75cd72e016aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-02 20:26:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cfc5b80c7335ffd30199bc1bb0997a8f2f16cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 22:32:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e5a9fa17dcd40a6811d27549069848a : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d0401b5a4c452389faf48df9e083d8 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9885266b0a24c699163c0b3930f584a : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8a5d1d6c444671a4e6ea50618c4e9d : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2917ac62eead4d4e85a8660c6905b76f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44ce15c4f48d4c3c99b49db430ad6030 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845daa983363451a849f0aac696dcd3f : SUCCESS in 14m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3841a83612984abab060c9e8a65eead8 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3fc345bab2af4b43adb05735c8a938c1 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/914cfce388f4479697f99176446387c4 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/17777359bdf844d1983978f149d6cebe : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c942375f4774a30a245fcbe4c9d6422 : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0c0836d257e4043a53d37e59de9744f : SUCCESS in 1h 28m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4c5752406874e769b01ebb502d0ceb5 : FAILURE in 1h 33m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/704b245d93ff4c5db3ddebde6c748fde : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/585173cbc04c4663be24fe35395623ce : SUCCESS in 35m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/019d5257fbd74b52a6d0fc5625d9c67c : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d28a3aa26c8144a5b93ccd080fedc09d : SUCCESS in 58m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/513ca1eb99bb4fa4ae3196a4cdf480a4 : FAILURE in 1h 24m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/887162c2781648d3a82170072529bdfc : FAILURE in 1h 44m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6ed7b044a1c428693df3a7797db6f26 : FAILURE in 1h 53m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/beef2294132f410b8ec4bb62fed83583 : FAILURE in 19m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7113ebc70a61477b9877a8b68de0f0e4 : FAILURE in 1h 18m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5efffbe0fb3d4bdba20584bdaa5098f7 : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa0f9bda2f6d775dc4c1ba5318ddc36eff15a188","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-03 13:50:48.000000000","message":"Patch Set 1:\n\nDoes this need a tempest update?  Seems some tests are failing to find a subnet by ID.","accounts_in_message":[],"_revision_number":1},{"id":"896e3158e6892c04099df09d1381d974f666f765","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-10 07:32:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Does this need a tempest update?  Seems some tests are failing to find a subnet by ID.\n\nYes, the fix breaks our tempest tests. Our tests use ipv6-private-subnet from devstack, a subnet from a non-shared network, attached to another project/user.\n\nI proposed a patch in octavia-tempest-plugin to fix it:\nhttps://review.opendev.org/c/openstack/octavia-tempest-plugin/+/774157","accounts_in_message":[],"_revision_number":1},{"id":"2717ec10e569b66c121a6d2699635d3ec58d8000","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-02-11 13:04:44.000000000","message":"Patch Set 1: Code-Review-1\n\nWhether to use our admin context or the user\u0027s context for validation of the subnet probably needs to be linked to the same config variable we added for the previous related security fix on the network_id side: allow_invisible_resource_usage. My current deployment (and the one I operated previously at GF) would be non-functional if we enforced this, as users don\u0027t own or have visibility to the subnets.","accounts_in_message":[],"_revision_number":1},{"id":"af5434a95b02945726d5fb2b34f8059e71fad9b4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-11 13:16:32.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Whether to use our admin context or the user\u0027s context for validation of the subnet probably needs to be linked to the same config variable we added for the previous related security fix on the network_id side: allow_invisible_resource_usage. My current deployment (and the one I operated previously at GF) would be non-functional if we enforced this, as users don\u0027t own or have visibility to the subnets.\n\nHi Adam,\nallow_invisible_resource_usage is evaluated in the _get_resource function (https://opendev.org/openstack/octavia/src/branch/master/octavia/network/drivers/neutron/base.py#L187-L189) that is called from get_subnet, so it would not affect your deployment if the option is enabled.","accounts_in_message":[],"_revision_number":1},{"id":"1dd611b5d2e5a1e4e97ce5448535e8568f0c2de8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-11 13:24:16.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Whether to use our admin context or the user\u0027s context for validation of the subnet probably needs to be linked to the same config variable we added for the previous related security fix on the network_id side: allow_invisible_resource_usage. My current deployment (and the one I operated previously at GF) would be non-functional if we enforced this, as users don\u0027t own or have visibility to the subnets.\n\u003e \n\u003e Hi Adam,\n\u003e allow_invisible_resource_usage is evaluated in the _get_resource function (https://opendev.org/openstack/octavia/src/branch/master/octavia/network/drivers/neutron/base.py#L187-L189) that is called from get_subnet, so it would not affect your deployment if the option is enabled.\n\nI did a quick test: setting allow_invisiable_resource_usage to True allows the user to use a subnet_id from another project/user.","accounts_in_message":[],"_revision_number":1},{"id":"0706a37a8b9d6237ecccfbaebca8858713b4d07c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-02-11 13:36:58.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e Whether to use our admin context or the user\u0027s context for validation of the subnet probably needs to be linked to the same config variable we added for the previous related security fix on the network_id side: allow_invisible_resource_usage. My current deployment (and the one I operated previously at GF) would be non-functional if we enforced this, as users don\u0027t own or have visibility to the subnets.\n\u003e \u003e \n\u003e \u003e Hi Adam,\n\u003e \u003e allow_invisible_resource_usage is evaluated in the _get_resource function (https://opendev.org/openstack/octavia/src/branch/master/octavia/network/drivers/neutron/base.py#L187-L189) that is called from get_subnet, so it would not affect your deployment if the option is enabled.\n\u003e \n\u003e I did a quick test: setting allow_invisiable_resource_usage to True allows the user to use a subnet_id from another project/user.\n\nAh, ok -- interesting then that it\u0027s breaking our tempest tests. Guess we just did them wrong. Will remove -1 for a bit until I can re-evaluate.","accounts_in_message":[],"_revision_number":1},{"id":"a9aa7347ca9adc8b87b8c97b75af6e377adba667","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-11 13:43:41.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Whether to use our admin context or the user\u0027s context for validation of the subnet probably needs to be linked to the same config variable we added for the previous related security fix on the network_id side: allow_invisible_resource_usage. My current deployment (and the one I operated previously at GF) would be non-functional if we enforced this, as users don\u0027t own or have visibility to the subnets.\n\u003e \u003e \u003e \n\u003e \u003e \u003e Hi Adam,\n\u003e \u003e \u003e allow_invisible_resource_usage is evaluated in the _get_resource function (https://opendev.org/openstack/octavia/src/branch/master/octavia/network/drivers/neutron/base.py#L187-L189) that is called from get_subnet, so it would not affect your deployment if the option is enabled.\n\u003e \u003e \n\u003e \u003e I did a quick test: setting allow_invisiable_resource_usage to True allows the user to use a subnet_id from another project/user.\n\u003e \n\u003e Ah, ok -- interesting then that it\u0027s breaking our tempest tests. Guess we just did them wrong. Will remove -1 for a bit until I can re-evaluate.\n\nYeah it breaks the tempest tests because we don\u0027t enable allow_invisible_resource_usage and tempest tries to re-use ipv6-private-subnet created by devstack (this is the desired behavior, so now we need to fix octavia-tempest-plugin, it should only use its own subnets/networks).","accounts_in_message":[],"_revision_number":1},{"id":"296c8181fcbb44e593262a05882001945179789f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-15 15:55:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fa5afa2e1dfc882c94b1c6abf6383be7c1366378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:34:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/187bf8b9416f413c82738c273e4bfae6 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/880b6fa8c0d24f85bef964705cd5e4a2 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec112a11603042b3b339165c1dbd8dc1 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54acd5fd6fa04fbba92ff23ad9f88f53 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8864b72e1ad34de2915883e45226d399 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56215ffc61c44d59b08a57814a08acc7 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7dcc4aeaef941f390b18f70e942932f : SUCCESS in 16m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef86688508de4bedbac9ed02713b0c48 : SUCCESS in 5m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e7935f78d2e4b91bb271dbc1d458766 : SUCCESS in 9m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a2976932f2e54bfd91368bbffcbb9981 : SUCCESS in 11m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be8ce2660c114bdd805b4e9514627828 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04652b402f31499c97cba0d6dec68505 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9cee487214f4e0ea3e82955b24268d8 : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/646ebbc83ee647c49475e1bf4b314ddf : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/042d179a36f7431db9574a28cde9b508 : SUCCESS in 44m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/21efc64b5857459eb8a63b100d90a139 : SUCCESS in 42m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3337e95c55641a79e4af9a12ed730f9 : SUCCESS in 1h 07m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8c7831216654399af8d3484da9a8880 : SUCCESS in 1h 01m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c098f04138742ea91a817708e4043ff : SUCCESS in 2h 04m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eb2f895d24f7499ea83d8b6038dbcaf8 : FAILURE in 2h 11m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc0580ad1864579aaf75e18914ce359 : SUCCESS in 2h 02m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c9bcbcfd84624729a153fe6ba4901f33 : SUCCESS in 2h 04m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fce64be4899c467c90e6c557c5313cc4 : SUCCESS in 2h 13m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/69de647f25764724ac579af3874d411f : FAILURE in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a67e254085c1df04663c783cb8c0d104151ea0a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-17 06:50:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11f9453e8c2ccc8de1969110d46f6864cd59ad47","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-18 14:51:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"00efa7d6abee371d9cd44e0e9e5080de3f71765c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 11:09:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f97cba9f6425a8f89d115af0ed907ec9cceb1d73","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-03-08 07:50:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"45e69bc855cc5d99c3d73f3faf4ddf400704894d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-20 20:33:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"cdd460f64a793b1a58e03b4040512b0e8188168f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-20 20:34:20.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5847dab0e480928d733a4931822416be4a96bb30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 22:28:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d54aa910d2b14220af4916c2ede1ecfd : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53c5a38b206142228cdb5d5076db5b8f : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46532dc10ae4b64923d90561f86c78f : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d9da5c0e21a40f1b3846e5c7cd92235 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7150e6a1a5b488bb298683eaced7e9a : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67abddb4da76450d90c011ca2c71ae2a : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd33e45e003493db347766164c8f2f3 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/087f3031865044109eaae895decc8e61 : SUCCESS in 7m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/98c04865003a4b509fcde45cb9dddb29 : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6923792cff846548912a327b7f387e1 : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f41a30843955423b93d0e0e25f9b5a45 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb1406c8fb7b4d3bbee4f1124068889e : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2a85829c3f746258158f5e19b4013f7 : SUCCESS in 1h 19m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51f5e0e992734b86a363c4fef95fdc99 : SUCCESS in 1h 19m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aff4951b92564c87bcc7ec159d4143f6 : SUCCESS in 37m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9365097065234bfabf98310b2ae9d702 : SUCCESS in 32m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/355250f5015f4c7f949fc5d3361cf254 : SUCCESS in 36m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33b025b6189c41718ae66df5d885c2d1 : SUCCESS in 48m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/21147cd79d164247a738c3f7f8bed4de : SUCCESS in 1h 20m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f2ebe8ed6e64369b80784a2e3ebaceb : FAILURE in 43m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ef7859c0a1014eb6835eddc5124c4327 : SUCCESS in 1h 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b44f86070a940b7bc962a31f58cda0f : SUCCESS in 1h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a8e4bba1a441407cb930f9d18caa7e42 : FAILURE in 1h 53m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ee61d2ea13014c18933a7a53d2bbb658 : FAILURE in 28m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a61f22f4598891ab13d9e097408a6d9b8c42a822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:59:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f014ec964ddaefd4b0846ccb22acc5de781cbbfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:43:17.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"b7c2aded2a8f294e50cb4d0f555d6bc980dcea68","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 00:54:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e973f232a86a896dbe014c4f7dc57d2925395b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 03:16:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c215ae915744493d8c62cee3ea1ff5a2 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f721df3367f143f7a32760e145303593 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d54398e4a154a94a8daa392f5b1ef64 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d028aed8dac4291921be0c5f577d817 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75fcace3e9a341958743fccb20f29e56 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c53e1783834af5915e2ea430e0374b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a37aa66d2f4f6e8ced05c96d43badf : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65725ba6dd774de992b5e4889296c8cc : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5d32e63dc56245878db1e119e23fdf0b : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05636fad75124381b622254c061b1834 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fb9683fb4df1434bb3aac18b008454b6 : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80840543aa114e788e43137b409f3b9f : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fde2270ad5ee4c6889bc628c42befdbb : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55212299ea6f4925b07d7512dfcffbd3 : SUCCESS in 1h 59m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d71b5b74da7482aa4d52981d399cad3 : SUCCESS in 52m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ab83310b286c4284b2134d18bd69887b : SUCCESS in 50m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4aebd9ab5d04d779ab3dfca99a66e30 : SUCCESS in 51m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da9c0ce3ed49480284ebd65c05884a62 : SUCCESS in 1h 09m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ac64d4f98c644ff9ff53c6447b3dad8 : SUCCESS in 2h 14m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2a01220ec734805b050d1ea210f831a : FAILURE in 51m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a3d6ae8165e4ef6a6af0c333d3f13a6 : SUCCESS in 2h 20m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a6174d56f354cb982c68c7aa30fc25a : FAILURE in 1h 49m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/62d73f3769d845dfaa5cdaed82d77cc0 : SUCCESS in 1h 59m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fac2bbbe39cc4ec5b4dbcf85d21b3161 : FAILURE in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c2af1c5219ffc106dee8e23f10395f63a0c96c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 03:17:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d994706bbb17018051016e6f0096eae11630bda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:10:40.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"adfb8f94e7985d1e1dc60994997644cafedd68a7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 04:11:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b9f84ba96c39f8129ea71a528afc2d5c06dbf968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:09:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c0e8287361a458c903129858e01989a : SUCCESS in 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a598342b795c41adb7e7bbc94df4cb42 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ada2a99e72d44a5a254eaf38a532817 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943b018e760d4ae3abe31e38c99d7fa9 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65e06397c3df474488f1f804abf9f44b : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c663db1893e341bf910114c809099f66 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50f6747ddda45ec9e338ea95348f466 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5e233d58f5e4504bb137625e25b7430 : SUCCESS in 5m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/525b68da31034c40975b82d4f7a9b50a : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/be159933fc7f4673a016a98be1a285d4 : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7d117067314435297dd61d748173ea2 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bd0aefa02f54c9e853b67b43bff8805 : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/881c3d9ff73e4d04a69393e1cec54721 : SUCCESS in 1h 24m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8912105ab44e4d10b1db389eba2e135f : SUCCESS in 1h 55m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/921c3944a0764036a05690b15925eb97 : SUCCESS in 45m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/01530aedb70741228deb52ea979c7a04 : SUCCESS in 41m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a9104751da443b9bea3e61ffb8aac39 : SUCCESS in 44m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f1a57b1cc604421bca15250e456703d : FAILURE in 20m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/039536fdba6d430785b7874a6e43221e : SUCCESS in 1h 29m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3620e43f2244affbaa5d8df79c48b43 : FAILURE in 1h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6897e73d88c14c368cea5a15e27a4663 : FAILURE in 19m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64ba6341fdfb43f29d860614fd001431 : SUCCESS in 1h 49m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7bbcfdf98d7b48c895e040202df09ed8 : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d57f02e9d3f6480ebbc8dcef36de8239 : FAILURE in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"67fedfc093f9b372af5d1bee666c53b5a794ace3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-26 06:16:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a27c9a2820138573ec51a6dcc3657cbee8dcb51e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 08:43:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eb9004ef98d46eebabae0a3b1ec5694 : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8f8c4b4320140dfade9d1b8599318eb : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295bf3d7ed84459bbee2166f49e9c664 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a9cda5933814e52b9e7c990693d016f : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2036e1d3c3a43e1ac80f3e34c1b4261 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60de3bc657bd4ab79bda2fa1cb236a74 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d91e778f2ef40c498e7853034f171e2 : SUCCESS in 15m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b00f1fffb3ae4c7da22beca6169ead66 : SUCCESS in 6m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8b7cfdc656b44991b53e25aaf472863e : SUCCESS in 7m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b564caa3b8b144a98be8e1b14f214a23 : SUCCESS in 10m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5066a8abb854c9e97eab0f512c25167 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4eb0b0f4e7ca4b44b13b1290f83cd030 : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a8575ed57bf4e1f9379b787b2b388a6 : SUCCESS in 1h 06m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20b2b01d0d3b4bfc8fabf8d47fa21400 : FAILURE in 1h 45m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d545ad1144ad4f7483c1c13f4d8820a2 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/615f56645e534381ad4f431b25686d7b : SUCCESS in 40m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23247d2cc8054a9eb748f16e5eeb20af : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/130c0cb28ebb4ca1b0e0325ec3a9dcc1 : SUCCESS in 1h 04m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50a3e088368b4482916cde1071cf4eed : SUCCESS in 2h 01m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6cbe4da151f344e79d53e48d35c40020 : FAILURE in 53m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0df85576c0fd49e5a74af044e10247c4 : SUCCESS in 1h 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5c059aa6854249bf87784e91188b2879 : SUCCESS in 2h 05m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f91809846b6b4a36b792f4a7d1a0c7f5 : SUCCESS in 2h 09m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2cab893260304daeb52ed309f0e9868c : FAILURE in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bdd3c4c5847afc1338d80068205fd038144ea727","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-26 15:21:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f7035de77d4ec491cd21cc32a2f80b29841a611f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:56:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afa2f9b83d6e4d82a0ef388ee12fc6b8 : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/321fafb2be6e44a6984aade2c64ffa10 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936f996694754281bf009f46f278d733 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3daa3c1b758c4940a516c8015f003853 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3162d95cb1854c6eb2bdac24f1413470 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5877d2fff5946c38a7ee88a45f5f277 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de45d15d3fbf437aa327ad3e78e3b161 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fb55508ccb6460ca22da1e5135a3408 : SUCCESS in 5m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fef991c255564f319ba9bb0da6487d8a : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1ed103a8a22e4854a8b0c7f066287bb8 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1195c9e8ba414c34b2018901eed241ce : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ad7e2ff5b3e47d9b9eedfd590787c09 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc6069a9ca434ea8909e40458a68b5c4 : SUCCESS in 1h 13m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c6fb9e6cf44486e9a1b6f211c1fa796 : SUCCESS in 2h 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e1f526b261b4f2f919718383aa8dcea : SUCCESS in 53m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d759a3affc444aebaca99fafb9a607b8 : SUCCESS in 40m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99a6746e51e8412eb3c61e00b09593d9 : SUCCESS in 41m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6e52e92cf7a485087f99c8bb30bc444 : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8f9762e0949347abace57037a5613614 : FAILURE in 1h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5426b6b9a3b7484593fd55ed687bf1b7 : FAILURE in 45m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f457971dbd44913abeeaab2833a3f74 : SUCCESS in 1h 54m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2760b14e700642ddadf4a7835ec7358a : SUCCESS in 2h 18m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5634d3ad07ce4691874c1644b7921098 : SUCCESS in 1h 56m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0997fd7b06924f3e8b2821d0ad9b3737 : FAILURE in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d92f0a75519b369bc3f76ed164eb38b2ef8a180b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 23:38:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e932876b9a14c30037f7206d72fd4686c1996042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 01:21:17.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"f013c54ca4559e3f022f986f9be6f5faf665fe61","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-27 07:46:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dd5471c5fc622f67a2082b13e7d7f4d872617522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 09:59:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a05bd0451d5246699357f125733cde68 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/916f791639b940e8a3f34f58d283c267 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef1fad483d5488baf339de073fbc278 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/babe45092d274edcbbc9ca32aa75105c : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f4c8d9aa9b245049b846a0ff091b1e0 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8846f0392474f4bb87bd059f6974467 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f193249181a945bc9e394f8853070f44 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/461c64e5ca944345841b1f541a47031f : SUCCESS in 5m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/805c6c548bab4c14b9c0710f3f599119 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/46d70dca5bf74509bdc1261c166602f5 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9f0bb5532034ba8a308b8fe4f36fc18 : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a9ee12bea9f47f3b0dc245a60e0b387 : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da3fdbeb1d7d46aebc2bb3aa72c7d68f : SUCCESS in 1h 18m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5834df421de1426f9ccb1681f9fce737 : SUCCESS in 1h 58m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/454cb41b5e3f4401a1a4fa22e4ae73c5 : SUCCESS in 42m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e525a6a9d2f64a67978b0d7c0221862d : SUCCESS in 36m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c28e94836514874a568deeb64d414b4 : SUCCESS in 55m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fcb6d3f689e14b2aa21848535645c4c4 : SUCCESS in 56m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/358474a8eec94cf59e03f092d52400ce : SUCCESS in 1h 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c02a88881db54bb68b262b295a8fbbe5 : FAILURE in 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9e455bc734794d9f992f5e344653cb87 : SUCCESS in 1h 58m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eee699fc57f24ee5bbe5f9d398e60182 : SUCCESS in 1h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8afc056887f2442da36d9e52e7652182 : SUCCESS in 1h 50m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/df913cb496e84c478afc52ef24bbbb2e : FAILURE in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2131f17923f079fcbbf4b0b88138e028f6c65a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 09:59:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ca41ace38d13a0a6122d9cf33de07ce4e9475e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 11:44:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69b04059226644e29bfcc952e115dff1 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760b41e7e76f49d090da4cf6569cfabd : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/493792f833e64fb3b415e27856da6d4b : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b491a5f1579142c683d7ce110c20656a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0978d63f3dcb4f2c9e2ce3ab0f0af2a6 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb1163392c0f4c8085b8207d16b021d7 : SUCCESS in 5m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1af6f6ea84e0400c8373b90ac55e77af : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/841c5130bc0a45799b3ef186fcee9117 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02331c6c0537423687c5dca0171f37f5 : SUCCESS in 1h 33m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75a2e740d63b410cb9eae612ad13575f : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/344528ae90c946e3986f825a8db97511 : SUCCESS in 56m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95ee4e5a2ef344dd9dad3f1e55a4f9ae : SUCCESS in 43m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2974bd7839904c21af3eb8da21275dbf : SUCCESS in 1h 05m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa5b96ef9cf041cd801dc2d82abd93ad : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":3},{"id":"38235a19400a665c3e4e6a9232eefceb03e03b2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 11:44:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"55172d284bc7d24b7f0b4e5c1c03b2946a75d36e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 11:45:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853eb80b8de0499185d8f9c21588530e : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97742fef63c242198b8715eee375e11c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8d86187c0abd282d10a9fbe294df68f9100ad4a0","revisions":{"f871ac6503cdb5a03a12137dcc18aef30281860f":{"kind":"REWORK","_number":1,"created":"2021-02-02 20:26:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/773798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/773798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/773798/1"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 18:58:14.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 20:11:59.000000000","tz":60},"subject":"Validate user access to vip_subnet_id when creating a LB","message":"Validate user access to vip_subnet_id when creating a LB\n\nA user was able to create a LB using a vip_subnet_id from another user\n(by passing the UUID).\nNow, the vip_subnet_id parameter is validated using the user context,\nso an error is returned if the subnet doesn\u0027t belong to the user.\n\nI479019a911b5a1acfc1951d1cbbc2a351089cb4d was a previous attempt to fix\nthat bug but vip_subnet_id check was missing.\n\nStory: 2008586\nTask: 41741\n\nChange-Id: I602418264e171a2b1a926eff0b1f9e6dc186295a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f871ac6503cdb5a03a12137dcc18aef30281860f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f871ac6503cdb5a03a12137dcc18aef30281860f"}]},"branch":"refs/heads/master"},"6c1537cec6426dc143f869755a1d3e4b47ef0b99":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-15 15:55:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/773798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/773798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/773798/2"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 18:58:14.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-15 15:55:05.000000000","tz":60},"subject":"Validate user access to vip_subnet_id when creating a LB","message":"Validate user access to vip_subnet_id when creating a LB\n\nA user was able to create a LB using a vip_subnet_id from another user\n(by passing the UUID).\nNow, the vip_subnet_id parameter is validated using the user context,\nso an error is returned if the subnet doesn\u0027t belong to the user.\n\nI479019a911b5a1acfc1951d1cbbc2a351089cb4d was a previous attempt to fix\nthat bug but vip_subnet_id check was missing.\n\nStory: 2008586\nTask: 41741\n\nDepends-On: https://review.opendev.org/774157\n\nChange-Id: I602418264e171a2b1a926eff0b1f9e6dc186295a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c1537cec6426dc143f869755a1d3e4b47ef0b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c1537cec6426dc143f869755a1d3e4b47ef0b99"}]},"branch":"refs/heads/master"},"8d86187c0abd282d10a9fbe294df68f9100ad4a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-20 20:33:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/98/773798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/773798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/773798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/773798/3"}}},"commit":{"parents":[{"commit":"6c54eab5b5d8fd5300cd5d62e23e6cadc1641636","subject":"Make /healthcheck cache results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c54eab5b5d8fd5300cd5d62e23e6cadc1641636"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-02 18:58:14.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-20 20:33:48.000000000","tz":0},"subject":"Validate user access to vip_subnet_id when creating a LB","message":"Validate user access to vip_subnet_id when creating a LB\n\nA user was able to create a LB using a vip_subnet_id from another user\n(by passing the UUID).\nNow, the vip_subnet_id parameter is validated using the user context,\nso an error is returned if the subnet doesn\u0027t belong to the user.\n\nI479019a911b5a1acfc1951d1cbbc2a351089cb4d was a previous attempt to fix\nthat bug but vip_subnet_id check was missing.\n\nStory: 2008586\nTask: 41741\n\nDepends-On: https://review.opendev.org/774157\n\nChange-Id: I602418264e171a2b1a926eff0b1f9e6dc186295a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d86187c0abd282d10a9fbe294df68f9100ad4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d86187c0abd282d10a9fbe294df68f9100ad4a0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
