)]}'
{"id":"openstack%2Foctavia~774359","triplet_id":"openstack%2Foctavia~master~I6dd8e99b07ff557674871cb503dece96a9df3ada","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6dd8e99b07ff557674871cb503dece96a9df3ada","subject":"Disable auto configuration of new interfaces in NetworkManager","status":"MERGED","created":"2021-02-06 21:28:28.000000000","updated":"2021-02-16 23:57:06.000000000","submitted":"2021-02-16 23:55:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774359","meta_rev_id":"a3223e2c82372c9a8f360d424e139a7f7fd68b36","_number":774359,"virtual_id_number":774359,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 23:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"all":[{"value":1,"date":"2021-02-07 19:13:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":2,"date":"2021-02-16 21:50:29.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-02-16 21:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":1,"date":"2021-02-16 21:50:29.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":2,"date":"2021-02-16 21:50:29.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-02-16 21:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 23:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-07 19:13:00.000000000","updated_by":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"reviewer":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"state":"REVIEWER"},{"updated":"2021-02-08 15:13:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2021-02-16 21:42:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-16 21:50:29.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"55e47926de5a7edc2b24f052cd59877ab4380860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-06 21:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c122c4f08deb658515ac3c392fb7fd8163b546ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 23:40:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5da0252e30a438d8a4e4462c00b193e : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5716c0b1df9f45a79515f72bf97dbdbd : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b32b74203a4f0cb5cd5a8b37263382 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59806147676c427caad1d7120c3fd554 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a66fc056b14cb3a032a2a500a12ef9 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90ac097b2e654f129e9a617a9984e5d2 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d82b3c30bb349fdbe5bab70b6bb5bc8 : SUCCESS in 10m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3bb039a519af49188893aba6d2812bcb : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4baa996246e4dabbfa169a0da6f2441 : SUCCESS in 5m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46241756b2f64be597746a8464d5bff2 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/02ecd1ee6a5348209571ef220a5fb75b : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50f74374886a46e08d23365d1f4c752e : SUCCESS in 59m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46aa1bb0c677478ca800b99fc2c11b66 : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbac6825674d4b079d94b6f53eb3f371 : SUCCESS in 38m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d6af231979144bfe87aae80b673b8077 : SUCCESS in 30m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/939fd83e9c8243f4a0b336930a624297 : SUCCESS in 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca6dac08ca824bb39d91c8e9c4c7a270 : SUCCESS in 56m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6c86223ee2a04caba4f28790e1fc8e65 : SUCCESS in 1h 34m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f570a3e7cfc4bfb9b8ccc64a17ecab0 : FAILURE in 2h 10m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6dcf76406bf94ff89785c7c41d54d30e : SUCCESS in 1h 15m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/96ee04996ea44666aff8553200953a67 : SUCCESS in 1h 46m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/39e27c5975e449ef8016db1eb17a1406 : SUCCESS in 2h 01m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dde31a9b4a7c4e63a504b48aa853b509 : FAILURE in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"931515fc6cbe670cbcfcee44bf70482f9a8b73e9","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2021-02-07 19:13:00.000000000","message":"Patch Set 1: Code-Review+1\n\nWe tested this patch in our test environment and it fixes the issue.","accounts_in_message":[],"_revision_number":1},{"id":"c64e606139107dffa55b6a9289790ef6ca106c99","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 07:37:23.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e We tested this patch in our test environment and it fixes the issue.\n\nExcellent! Thanks for testing it!","accounts_in_message":[],"_revision_number":1},{"id":"e35206862ded0c6b5f9fc9033ab7d4f9abe4febb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 07:40:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eed3a061e83dce341bdf842359d81d3fb6e9777f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-08 15:13:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79946cba9c5f07c44af8a31c26d913115706aab5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-08 15:24:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"488cd5a7819199f09c7f2e75a9b4f926125233e0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-16 21:42:43.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"327a1cf0433c34543478a5305dcc3e1847f5f074","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-16 21:50:29.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05c3827564c3a8085519d8ee19abe35da77b0a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:50:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9af46b9be78cb39771b78918e9a9e0deab5954d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:55:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/307e27d0c58545ffbe56b7c6cca83ec9 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca202bbe2bef4b26b60a8c1e045d519b : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19ca8ac3a70c4868a099700fd78fd662 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0bae8d33be9439eb8654954e037b242 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aca9e0a90e4690a69209de698410c9 : SUCCESS in 12m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4a3b8f3f2fb42f9b3cd226bf7107c65 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eeb1d52ebd0a4ad3954bce91900631ca : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0af3c9536a9649f9a43e5f6cd126607d : SUCCESS in 1h 24m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68d69592953a4f538a31d268872f0446 : SUCCESS in 2h 01m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f77fcffd0894d488612d33363aee85c : SUCCESS in 55m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/049f16b693354358aec63166b81595ed : SUCCESS in 36m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b3321da3fae4c27a7d6a8df906740d7 : SUCCESS in 38m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1aeda2247614529b83e1f846c82e371 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"451e4afbcb090e58703845cf2cee0d0f7befdc19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:55:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a3223e2c82372c9a8f360d424e139a7f7fd68b36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:57:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3023969372741c08dae9a15882cf67d : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a518cefda1332042d50e5ce2814f0af933c62457","revisions":{"a518cefda1332042d50e5ce2814f0af933c62457":{"kind":"REWORK","_number":1,"created":"2021-02-06 21:28:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/774359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/774359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/774359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/774359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/774359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/774359/1"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-06 21:26:07.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-06 21:26:07.000000000","tz":60},"subject":"Disable auto configuration of new interfaces in NetworkManager","message":"Disable auto configuration of new interfaces in NetworkManager\n\nNetworkManager in Centos images configures new network devices as soon\nas they appear in the default namespace, it means that we might have\nconflicts between the management interface\u0027s routes and address and the\nnew VIP or member interfaces\u0027 routes and addresses during a small period\nof time before they are moved to the amphora-haproxy namespace.\n\nNow, the \"no-auto-default\u003d*\" option is enabled in NetworkManager, it\ndisables the configuration of new interfaces, while the management\ninterface is still enabled/configured through cloud-init.\n\nStory 2008599\nTask 41773\n\nChange-Id: I6dd8e99b07ff557674871cb503dece96a9df3ada\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a518cefda1332042d50e5ce2814f0af933c62457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a518cefda1332042d50e5ce2814f0af933c62457"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
