)]}'
{"id":"openstack%2Foctavia~774596","triplet_id":"openstack%2Foctavia~master~If2841720009c2e402127e2e0080efdd56b68f6c9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"If2841720009c2e402127e2e0080efdd56b68f6c9","subject":"Configure rsyslog on Octavia service nodes in devstack","status":"MERGED","created":"2021-02-09 07:47:12.000000000","updated":"2021-02-17 18:48:03.000000000","submitted":"2021-02-17 18:45:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774596","meta_rev_id":"6ca8f5a0768f546a535852e91a5891898c0c7626","_number":774596,"virtual_id_number":774596,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 18:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-02-17 16:35:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-17 16:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-02-17 16:35:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-02-17 16:35:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-09 10:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 17:01:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-17 16:33:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-17 16:35:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"e042493769b7c58b57f649b2b3f716cef843283a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 07:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d7135cdc184794554a4fdfc89c4a19092e554f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 10:12:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63b18d34bd684177879dc9b93e157dfc : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/452ba6a100b44169a02b2419ea9e5776 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a796b60d1294204bd8046b89d67db34 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c916bd6953e413d949bc665fbb54a0f : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb149fa34d94a9daa0b8652589f80d1 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/859ee776e20d41a89138b007ed6c3a1d : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3f3766dc964ebd885edd17b4fa313e : SUCCESS in 11m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47c3a926331945a39159e5337ab584ec : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e93da9c469264bd2ae9511c94c3f6097 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc4178c213344d189bf20ae4aaad7d27 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7d5440652464891b357e8d989901542 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/498948c5b4ac4595b2cc8489813215c6 : SUCCESS in 1h 12m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d28620ff0c68487c8e4912a32739fd13 : SUCCESS in 1h 55m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02cccd8edc5a4eb0b938edea88182ed0 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dfd959e7c6a64b8eb96acb038ef0dff8 : SUCCESS in 35m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43c58cada8364a4da6259885860f458e : SUCCESS in 38m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c078e5b84d974e7e86efb18c8cb15910 : SUCCESS in 57m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5c7477da29514c5699de31ab9e073af4 : SUCCESS in 1h 45m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e290894b996b4980aed6e1dc8560d27b : FAILURE in 1h 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09e7b679ee4241c8b0622a97d1d90e3f : SUCCESS in 2h 01m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0487e1b1d1a6427ba4bd44398445a934 : SUCCESS in 2h 05m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4d6218cbd6d74be0a5233a54c30b6f0c : SUCCESS in 2h 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c951406ae2884106904ece30d2f44788 : FAILURE in 39m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2026dfb4a48e6dee9e2f5fba4963b2838ad597c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-09 11:06:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c288de91107607b3db187f67a9cdb605afc4286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:43:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b82d6cdb227470fa072027e546fc468 : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eed7769e531e48a8ba30c85ce016a9a6 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2246cf8ebe92455c995c9e256a9402cb : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2045de9114fb4578b2b98772ee7a2c68 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8917970190fd483fb99fa34c1afd82b5 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ae309e1159041ed90522337ce37eff7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af8f2de8f3f4fe685f1846efa75604a : SUCCESS in 12m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6e1c79ea0d494a2c9b331250fac6ee92 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb6985f2cc694eac8796a12a86fa1eb4 : SUCCESS in 9m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/518ba956ce81462b91f831a7c04d42b2 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0890895202344641b4c11978468d44ea : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4efd644fcc34391895192a7fdadb145 : SUCCESS in 1h 13m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c13c39fb547f4999a084dfb198e721ba : SUCCESS in 2h 26m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8b08caf5efe4c58bd512e62de7ab6ce : SUCCESS in 40m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bd252403e4aa4527905d580da6a36233 : SUCCESS in 44m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5b23c95ce1a459c9adcc0424d25ab3b : SUCCESS in 41m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3663e705ebc64be9a90d639ffc66ad07 : SUCCESS in 1h 02m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab2f147f81fd40a687fcf3b5d0437547 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a73eb6a326c04de38475c69b9532583f : FAILURE in 2h 30m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6298654efada4848aa21ee25075dfee0 : SUCCESS in 1h 28m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a1288b917ed4b2b88e449283fe73a6d : SUCCESS in 1h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4fc012ade9724963a6abb0b65feb1219 : SUCCESS in 1h 43m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/abd6110ed0f9461a8cee63f044577696 : FAILURE in 53m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9dcffdcf1715d627a7dc51e4af8ce2148a36925","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 16:33:08.000000000","message":"Patch Set 2: Code-Review+2\n\nThis looks like an ok workaround to me.\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"7d4343d45f8265d42ca71b17adf0f7febc2603a3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-17 16:35:49.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35139aeb332b4c746a847d929bd5b077a9135dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 16:36:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2e5f3d342614ae65eb8839db7fe41f09e9cb6cf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:45:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7f95085867d4e52ae87a49bc85ad96a : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1bbc161c2146ccbe113ff4710b41c5 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721c6d943c994ef9b46b3542876c42a4 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45d9192b918b434786f0cc68bc4d1283 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e270ae1eb448f2b57b5dd33e2f93a7 : SUCCESS in 11m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cc4f5277e2504d09abe0cc9c0dc7813a : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6beef763752e4f7c8af22114d24d85da : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c2b8284f4b642548b5a77702b959689 : SUCCESS in 1h 14m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf87a0e4030541d28009fdc073cda0e8 : SUCCESS in 2h 05m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ca289eb90af4b1596a42fa1a2bb1751 : SUCCESS in 55m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7bedc95718304a1a8a153ddb667f635a : SUCCESS in 35m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd8f47388f194bab88509fb4e8848410 : SUCCESS in 39m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bced3b65480486c92e12e5efd976439 : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":2},{"id":"772f8fbe39d190657927721b5c9fb8d9180d20b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:45:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6ca8f5a0768f546a535852e91a5891898c0c7626","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:48:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8612e8b11248108b925e0a0734ea1a : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f45092a87673b54d578d8df058917e325d9b77b8","revisions":{"acab6c1ccdbd4fc0eba9de29d03da0cd7c6163ef":{"kind":"REWORK","_number":1,"created":"2021-02-09 07:47:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/96/774596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/774596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/774596/1"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 07:45:33.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 07:46:59.000000000","tz":60},"subject":"WIP Configure rsyslog on all nodes in devstack","message":"WIP Configure rsyslog on all nodes in devstack\n\n{admin,tenant}_log_targets options are configured with\nMGMT_PORT_IP in devstack, which contains the IP address\nof the local management interface. In multinode setup,\nit means that the second node should run a rsyslog\nservice to receive logs from amphorae that have been\nspawned by its worker.\n\nChange-Id: If2841720009c2e402127e2e0080efdd56b68f6c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acab6c1ccdbd4fc0eba9de29d03da0cd7c6163ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acab6c1ccdbd4fc0eba9de29d03da0cd7c6163ef"}]},"branch":"refs/heads/master"},"f45092a87673b54d578d8df058917e325d9b77b8":{"kind":"REWORK","_number":2,"created":"2021-02-09 11:06:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/96/774596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/774596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/774596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/774596/2"}}},"commit":{"parents":[{"commit":"9c6eb4917296b86bdc55c90b656c2a0979be2870","subject":"Merge \"Add default value for enabled column in l7rule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c6eb4917296b86bdc55c90b656c2a0979be2870"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 07:45:33.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-09 11:06:19.000000000","tz":60},"subject":"Configure rsyslog on Octavia service nodes in devstack","message":"Configure rsyslog on Octavia service nodes in devstack\n\n{admin,tenant}_log_targets options are configured with\nMGMT_PORT_IP in devstack, which contains the IP address\nof the local management interface. In multinode setup,\nit means that the second node should run a rsyslog\nservice to receive logs from amphorae that have been\nspawned by its worker.\n\nChange-Id: If2841720009c2e402127e2e0080efdd56b68f6c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f45092a87673b54d578d8df058917e325d9b77b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f45092a87673b54d578d8df058917e325d9b77b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
