)]}'
{"id":"openstack%2Foctavia~775419","triplet_id":"openstack%2Foctavia~master~I8ea9581364a822f84fd1c32d66dc6d55f00949c3","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8ea9581364a822f84fd1c32d66dc6d55f00949c3","subject":"Fix centos-8 job","status":"ABANDONED","created":"2021-02-12 17:02:13.000000000","updated":"2021-10-15 14:15:20.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e18efe112fcfcdcb21175e90f0656b3a937727ad","_number":775419,"virtual_id_number":775419,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-15 11:17:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-1,"date":"2021-02-16 12:26:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 17:03:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 20:12:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-16 12:26:15.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"160be2141de87a77fbd52fd712587f7246626be1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-12 17:02:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15eaed688c8aee52aae8c410d50437590d0ac037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 17:03:00.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 73, column 31","accounts_in_message":[],"_revision_number":1},{"id":"341ce2269d790ffa2a8acd85ec9552efc86507a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-12 17:50:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2487047ba5852093e34b0f8f50c683a8322cfab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 19:45:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a055f701b2d4e7c95bb30bed1f8d3db : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fecb7bfb709f46d7a760d67dbe0dc233 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a0283546f84493bedf9bb6cd06ff33 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538af1ec783a445390e8bf087975f0f2 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaba6417e0c348edaffbfdc11fc52a58 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a5d268deec349238f6b2931170dce44 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a58a46de3b9c401b9e052534c12af748 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f68c7480024a4d43bd68dff5d7c2a979 : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6d22b2af32444c49af396254f34cad83 : SUCCESS in 9m 10s\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0458f824a93c48ec999bb5ad0b5a07ad : SUCCESS in 1h 45m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"473069cb720dfb0ce2c74c87d43e1c7e48cb444f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-12 19:51:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"822a08c04d2ca74e2c373e0cff43e46827241702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 22:13:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3d90ef92d64fc1afaedc53d107fbac : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a67163c61c664651ade47d086aea9c5a : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0267e619c0824df1809005bf9b0bdd0b : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5785869b33f343da9895cd8957e5b05d : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47e4ac3cd5404bb99f8dd3a6eba44b33 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42a065e9c40d41b889f2809e4d43da26 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6262a6c480a3478484f74960dda9e7ff : SUCCESS in 12m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/84ab1a3840054f77b94c04d6e4a2ce2f : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40062f3ea6b74395a8e291e04e60a44f : SUCCESS in 8m 56s\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a84d6dbd0e8c4894a920e3972fc797b6 : SUCCESS in 2h 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"86133f5c9667c9b54e56dcb50e88122479ae9bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-15 08:52:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45976024b91c20cad309e169978a68d9344ab70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:17:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/baae17246ff843808eeba764b69eefe0 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ce245554f54b688af6b150df46a681 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01239d21a6f4372a0b43b100b9a0254 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f64f40df49c246cbbae608b87c9d4336 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9662483f8c24144960c5bffa1a78de0 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7adc151a4eeb44e5bdd9db80e0977615 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27793744184743e294e7d4eaaf47cd79 : SUCCESS in 11m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/117c9e020c274ceb99b50caf3d30bfe7 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/59c433d3f2b946d58154807a2f11abaf : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed4cc7b336574a3d938818e020e3d3a8 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e044f2628f14f9eaa0a008a4f6acb43 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ab18338f21e4817b7f765d9092bec9d : SUCCESS in 1h 26m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae4af94f0e6e4aa08d39e9bbfd071ee2 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/877ebedd00bf4fa596971afb9ae5553e : SUCCESS in 34m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e9edb77eed4e4067a90ff34d9861cbcc : SUCCESS in 55m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/def50ed45b14474dbd424b914bfbe231 : SUCCESS in 42m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ff1f6c1dcf74278bed7762a07def208 : SUCCESS in 1h 14m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41156ba0a5454d138257aeb01cb9c4ad : SUCCESS in 2h 04m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f107da5622204594b19cc836a8bc1671 : FAILURE in 2h 07m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/613861a9eaf74e32be408f47448cbb28 : SUCCESS in 2h 14m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/321749bcae31430892b3db3237ed94ff : SUCCESS in 1h 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0796ddec6e924a10a1e2695b80beb4a1 : FAILURE in 2h 22m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/28af2ad7343342a4bb9ce319a1481070 : SUCCESS in 1h 48m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b4a2b888f27f9e8b572b3058d9d203e7ea68bbc5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-16 12:26:15.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e18efe112fcfcdcb21175e90f0656b3a937727ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-15 14:15:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5f47c1bbc754c84f13fec95987102fd817ea55ac","revisions":{"47d1735f308447476856bf75a3c22175ada2f1bb":{"kind":"REWORK","_number":1,"created":"2021-02-12 17:02:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/775419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/775419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/775419/1"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 16:55:54.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 17:01:45.000000000","tz":60},"subject":"WIP Fix centos-8 job","message":"WIP Fix centos-8 job\n\nChange-Id: I8ea9581364a822f84fd1c32d66dc6d55f00949c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47d1735f308447476856bf75a3c22175ada2f1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47d1735f308447476856bf75a3c22175ada2f1bb"}]},"branch":"refs/heads/master"},"b095b283f42d4e2f1f3997ba53f0db65866e665a":{"kind":"REWORK","_number":2,"created":"2021-02-12 17:50:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/775419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/775419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/775419/2"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 16:55:54.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 17:50:49.000000000","tz":60},"subject":"WIP Fix centos-8 job","message":"WIP Fix centos-8 job\n\nChange-Id: I8ea9581364a822f84fd1c32d66dc6d55f00949c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b095b283f42d4e2f1f3997ba53f0db65866e665a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b095b283f42d4e2f1f3997ba53f0db65866e665a"}]},"branch":"refs/heads/master"},"040c9d92746e9b755ab1dcb76a08c897f29b6297":{"kind":"REWORK","_number":3,"created":"2021-02-12 19:51:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/775419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/775419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/775419/3"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 16:55:54.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 19:48:45.000000000","tz":60},"subject":"WIP Fix centos-8 job","message":"WIP Fix centos-8 job\n\nWorkaround issue with dib in Centos by using a recent release of\nsetuptools from RDO\u0027s RPM.\n\nChange-Id: I8ea9581364a822f84fd1c32d66dc6d55f00949c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/040c9d92746e9b755ab1dcb76a08c897f29b6297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/040c9d92746e9b755ab1dcb76a08c897f29b6297"}]},"branch":"refs/heads/master"},"5f47c1bbc754c84f13fec95987102fd817ea55ac":{"kind":"REWORK","_number":4,"created":"2021-02-15 08:52:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/775419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/775419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/775419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/775419/4"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-12 16:55:54.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-15 08:52:43.000000000","tz":60},"subject":"Fix centos-8 job","message":"Fix centos-8 job\n\nWorkaround issue with dib in Centos by using a recent release of\nsetuptools from RDO\u0027s RPM.\n\nChange-Id: I8ea9581364a822f84fd1c32d66dc6d55f00949c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f47c1bbc754c84f13fec95987102fd817ea55ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f47c1bbc754c84f13fec95987102fd817ea55ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
