)]}'
{"id":"openstack%2Foctavia~775957","triplet_id":"openstack%2Foctavia~master~I4443d4531dc97d14f8277024baa11ab43e87fb39","project":"openstack/octavia","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I4443d4531dc97d14f8277024baa11ab43e87fb39","subject":"Add support for scoped tokens and default roles","status":"MERGED","created":"2021-02-17 02:15:22.000000000","updated":"2021-03-10 02:56:25.000000000","submitted":"2021-03-10 02:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":15,"has_review_started":true,"submission_id":"775957-secure-rbac","meta_rev_id":"68a8d6799e57479b19ee1867577aa1f72d760b62","_number":775957,"virtual_id_number":775957,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 02:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-10 00:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-08 20:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-08 19:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-03-10 00:50:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 02:22:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-17 02:22:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-02-17 02:22:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-17 02:22:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-17 02:22:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-02-17 05:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 14:11:16.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-02-17 14:11:16.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"}],"messages":[{"id":"673783c04e5100a151ae8112d4d03873384a8db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 02:15:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bffa0c05c7fcb4c39396fa0528d4b27179a162a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 02:16:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c8968949d405e2cde1a2fbe248ac5705c52d95b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-17 02:33:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19b18a6a7b65af9883a79fd74207fb240c30a3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 05:10:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/465ee5e8e9674cc09ebfb4dc1eca597a : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd3290b497b744f682798b8d66a1eaaf : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff3e290f702f4406814eeb15bc71722b : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1ccf60bd1046979bb759e3e20d938e : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f6078ec60924f1388cd1e0251d9fda2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54208c68d6b4493ea7f35011977add57 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1450a3877af24a03b49e3aee09a1864e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644bc024592a4014bae0e20dabe57339 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3437d00d5ff24325bf6001392f33ed3c : SUCCESS in 4m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b5617ff29f9343e79f20d7e4203fc09c : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bc86cfb188be4f52a95dbe09257cdd8a : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef6f2d33360944218935ca05fb2adbe5 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/100d1acbe6c44fcbbe509d38d60858a8 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72dc7bcd27b94624ab4f278f16533f8a : SUCCESS in 1h 12m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad8e133d350743eb83a6488f99f1b560 : SUCCESS in 1h 23m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/432829b4823948c99206b3e525f7960e : SUCCESS in 1h 02m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0f5d68c01c474f649b9e7f05c00aa781 : SUCCESS in 35m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99d42cf71a114433a4f74ac126a4e9cc : SUCCESS in 39m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff29b6f70f9a4889b378f559a3fda42f : SUCCESS in 1h 27m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cfd0f2730354c1a8c8a0294db2a4393 : SUCCESS in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49f4757ce8134bcf821c7ae71b3204d8 : FAILURE in 1h 41m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4bdcc68383c04ac5a24ad7c549b3383e : SUCCESS in 2h 26m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49b25010672e4826ac7bb05d2bc14f80 : FAILURE in 1h 43m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7f05d3a2cf014181b4b12727dd50dd1c : SUCCESS in 2h 26m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/666f976d497a493992a3a66bae003318 : FAILURE in 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5b9dde89bc907b190178d2d12059ded574082ad5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-17 06:14:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fcb406d6ef38a04db927e94a496ef6a3fd9a3e09","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-17 14:20:36.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nThanks for the update, Michael.","accounts_in_message":[],"_revision_number":3},{"id":"3b28a583d2dc2ae4416bf29ceefe31beb0d5ccf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 00:23:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3eee2e2f8b863961b3432b06109279b1700fe798","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 00:24:04.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"437245430b8ce8ff81fdab287a59642baddfc3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 02:37:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/019e04cbbd91432d8014ed20fae47882 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24b9e47f094a4fb68bdb5726db54cc86 : FAILURE in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d977576a36f8465bb39c4f06e67c5e1d : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9d0826a5b04b1f87dc59c0210d3f9b : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3cd9570bba43909de1028f6ac3fb86 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2c549b2e384fb2b3f83f26c97e0ac3 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9644796091a34090a05dd2c75fdcc3e5 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7060af94d3f47f7840b13820399da00 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47a584b2c1344d2d92442beea6765744 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fd1fd530101d4c92aab6ed89e4c83d92 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/569063e8906f41d8bdff37df11c519d3 : SUCCESS in 6m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/989b5b54d5ca471eb7a361d072a7efc6 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7bb4137fa7bb4220a6145a574010773f : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd0fd60d6c9042a1ae4ed65bb151f3d3 : SUCCESS in 1h 12m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/058b3d1b36e84b1d8d24758c92b48e57 : SUCCESS in 2h 00m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ed248efc756481ba3344b4f6bad313e : SUCCESS in 40m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5a853f06510648fb857b8c2b6e0c1565 : SUCCESS in 40m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3d3d59273fa4474aa2e67f9512ac8d1 : SUCCESS in 40m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10b0f1d13ed04e03850b16f6fd920827 : SUCCESS in 1h 02m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80a7d34d5c9f4202b33049fa6d0d2282 : SUCCESS in 1h 53m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a724046ba8b0416e9a245525c60505e8 : FAILURE in 2h 11m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d32c65f09f804b72887990ca57bcc8b2 : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/69798d21c4384ba9b96c60054a85d5f6 : SUCCESS in 1h 47m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bbac729fada94df6b87222fca92884d4 : SUCCESS in 1h 47m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ab34e9a34dc54cddb339d3afe74a9f07 : FAILURE in 30m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a909bb2ab6ba85287e4462c186a19d2ac1669149","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-18 14:57:11.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"21d6205a3be92f169ef1d16f2d18f6bb4436032d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-18 15:18:27.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nNice job :)","accounts_in_message":[],"_revision_number":4},{"id":"581df50d827052a331fbfc2250e0b35f0555156e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 23:10:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e61fa5bb7e6af04093a6c34879d9d0ebc2eb316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 23:17:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e991c88a2a52493885dc6fbeddc422e93c9c1d27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 23:32:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ad553e83c54fa8497f50bc6a378efd1cc5035c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 23:39:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d2539553297fec4671fc0558eba805b7d03e3baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-18 23:49:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"99d3f854765e5b6c81dc3dd46bad08641adc24aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 00:07:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3b106afd13711c47ad96e645a33cc37ea0c0d816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 00:09:27.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"96c5e243543b418c12c1cd60a9e2a1274fd00acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 00:10:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6e5fe68fd81de2db1bdb2f4c111c5f7812f4b6a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 02:28:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cbd9556fe824a7e9aa32c1f29b11b95 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed5147d526a4cd1a867df538cef3f0c : FAILURE in 8m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b617d1643af436f811261b7e58f9bd6 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b7faa9b06e46f58a691977e46ca08d : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ada5360a3b3e4f608b646b4419c65412 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2aee0f716bc40c5be3f54c8e6230865 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1692130bec68468996b6f6085a04a6e8 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e840bb6551e43488e3cdec4fa034a36 : FAILURE in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02c62b37cf4a4ebc8ebff774de30f87e : SUCCESS in 5m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bac628ee35074805a10fcbc545dfaa11 : SUCCESS in 6m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6677203f0d0846eab9230befc965507f : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c23347afef4c46f3a45119d2b6c7c237 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbe1a12127894798a8a51024afc7ec9b : SUCCESS in 9m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3dd624570a4046f68aa54e7d728bea92 : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cba1b4f2e6a4ed58defb7c1c242405a : SUCCESS in 1h 38m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/582d91e1b33e4879b51bbffeb87e5c96 : SUCCESS in 38m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c77725e0476044a1b50648215fa153c6 : SUCCESS in 39m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18ed45f539a54932a5be380dc0019270 : SUCCESS in 36m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78d2126553a0441bbb5d7422f422b96d : SUCCESS in 1h 10m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/396a77c8e74945899c0fe348ac4e0242 : SUCCESS in 2h 11m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1bab507401b14c64a2087160716b245f : FAILURE in 2h 15m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf313b06c64f4a78a9dc3cb1dff4fab1 : FAILURE in 1h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c02c368bd7d435e889ed316146e95c6 : SUCCESS in 1h 39m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f53992ee12314345afc72f35b3d8e2d0 : SUCCESS in 2h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/776cea50b2304ab39a258b6bf00d37da : FAILURE in 33m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/install/install-ubuntu.rst","accounts_in_message":[],"_revision_number":11},{"id":"4e90eb456027d10a11e85965442209ffe6fd35f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 15:55:14.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"ab0622123bd3532c0ed6953238b501f2bb987fee","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 17:21:56.000000000","message":"Patch Set 12:\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b1f45f8e047e7e60f030130131f0c9df478a52d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 17:50:10.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"87e698a472648520ddc844d100d112b78bfc214e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-19 19:02:11.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cd2d01292b12c63ca0acb9777726d6e885d7d9bb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 19:05:26.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"de45db71e501c8daebc85c4fe9f89da6c3d16e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:20:41.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7a84cf4d2dac49c49f039469f10cf638 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afedd2c81fb8446b8d49f6e37ffaf560 : SUCCESS in 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2db7701fa4d34b3e95d2ecd41e406497 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bc34b8f501b4218b8bd394fa305935f : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e59c93d22e443f9b38a459d10bc72e3 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bac320b8dc0a476eae83b7370b1b4c36 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b88e719aa244328b58fe0c7b9fa506 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697d1e7e820f4e1c8e65e96f36493bc5 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e13247a698140319d10e1cd5fd5c4cc : SUCCESS in 5m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95de1272e89c40bc92a331d4824886c7 : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/be892b3fca0043f3aad8dc7d0c00303a : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef89557ae1474ad9be1571911c38742b : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca462802275e494da99a39b503bead12 : SUCCESS in 13m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bfebf01cbc8433f93a9d83378ad7f9f : SUCCESS in 1h 18m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fed8a6a0cdc4419982e7598817226591 : SUCCESS in 1h 59m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57c4f06ee46d4e4cba2c745e8bde8f70 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f6da81acd344488b90f66d0dd90214bc : SUCCESS in 1h 17m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9652cd5609c4daca64ab2f0c67ec68d : SUCCESS in 38m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c154d99f1e78425ebb3a7636ba134c9c : SUCCESS in 1h 02m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b2a9b3f53a884d6aa7fc14f8ccf148fc : SUCCESS in 1h 46m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36312fa362114dddbf04407715f5f2dc : FAILURE in 1h 54m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4adbe03042894e48861df19c473cd842 : SUCCESS in 1h 37m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8253203e0b7a405cac1a89d7b637411a : SUCCESS in 1h 47m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/053d00ab96d842d39661e4299d98bf6e : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b28e3c9ffd3c4086bb08d6b5a02603a8 : FAILURE in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b6ec9fc559e952fb7c4b55733c80298e6495dac2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 18:11:17.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ce6d036f017defb8332dbdfc605b40433375d618","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-23 07:45:40.000000000","message":"Patch Set 13: Code-Review-1","accounts_in_message":[],"_revision_number":13},{"id":"4b285f944e564c052e2d3032fb2209dac1dad066","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 14:26:00.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8bf256ca9c3c76ba237da5bd15bac8e79fe042ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-23 23:25:11.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4887b3b8962048c65d76e69c42d204916467c25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-23 23:25:46.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"9827bfdcfa7b13e81f7c65f6774f9ec23f079b3c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-23 23:26:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e02232f819c160a344f6ad11ee81beb54b333603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 01:47:32.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/36a8d9886e004344ba8a617655c434ab : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044af865de664eb2b164bddce270c04b : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35cd6aca83214d2297cbb73fc89ece42 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2c2cc1672e4709947e0b0eef535631 : FAILURE in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2e4d03ca974306bf1ecca28d8fcfaf : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314c10dde78e49f3b233ac226f277852 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba581381e42f47e48924266537c2ed5b : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ff22eb8bd14d38990f1f567dc39b63 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aadd5bde8a94d308ddedd1dfe7046cc : SUCCESS in 5m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7150e439f231490a9c74b1748aa485e9 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/811f78d0729c4a3fac269f06a9b7bc31 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1afcbb1a57c040099c88984c9c36859f : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46b9984f92e24f37a191e7e75c5c6571 : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84d012e9879d4d24bd6e92bc074a6aca : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb83c22de7b6421abb8a1010aff7d397 : SUCCESS in 1h 57m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/844134d4a29f411c8c2f180c5398ca77 : SUCCESS in 42m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bbed2c3689014f1887963adbfe3b944e : SUCCESS in 40m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e88a1515acc4f628247413bd9845e9c : SUCCESS in 51m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff82f0c4fe654e6ebfb650ff4cd81967 : SUCCESS in 59m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1c29c64a5c348419d423c46cf019524 : SUCCESS in 2h 07m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2326e8cf86984f10b0da0f6bf72466a9 : FAILURE in 2h 10m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e46c2284dcc14b919af2e0cbf3934213 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc6af8271e7e477a9ec4fd2facaacc46 : SUCCESS in 2h 13m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a8badc67cb04f65997c9944bd390463 : SUCCESS in 1h 45m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f05556e55e5f4bb5a96e288eb6f81dde : FAILURE in 32m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/common/utils.py","accounts_in_message":[],"_revision_number":14},{"id":"594e59c85c8f440ed1fcc97eba5e2e5029b2309d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-24 10:32:23.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e2ee52c88b658df45d33be2b808c5f71b5c9b940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 15:38:18.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"06798b98c205cceec4c31166d0478cc01ee4eed5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-24 15:38:47.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"1ce529703667c30cbe3bc14a6300c628c0e36998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 20:59:09.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/32b5ed3aa0d94814bed1f008c423c963 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72d4debfa19a4e10b014f1accd7b17fa : SUCCESS in 7m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ca1017bd70b4bc3b7e73ccfa5ebb183 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ee0f7517f1412684740075ea501303 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6be83d00df49d68c27b1e7a18fbe5b : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e147bb954324ab58a332e22967936dd : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/903e6c29a136444aa92ec826949e213b : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6453475953c40b0b1466155ae2217e3 : SUCCESS in 16m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70df724a42c449dea9b70c2df69108fc : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9da4776c6cbe45a985e6298ecfdb0f86 : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9e18e9e1edb453ca4c6bc2bb7eb6ead : SUCCESS in 10m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e19226c18ef24d0a8cac547d44a9ee2a : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ac494067a6c412c956e0f2fb236f7ad : SUCCESS in 9m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/907bf996125b4b8a82b44e95937184ae : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56442f5db6444ac9b7720ee816851ca2 : SUCCESS in 2h 16m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3d8240c73874fb284a5ac0a21252dd5 : SUCCESS in 40m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1249a3fc8be24c01a077188331497da3 : SUCCESS in 38m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/544cb899b40a4ac79eb0e447a3050be7 : SUCCESS in 36m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bf7015726ac405699f3786a3ef0579f : SUCCESS in 1h 03m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8072ff1263e843649771c2b0d74d865c : SUCCESS in 2h 20m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/062cda0b84c645739120d40b6fd82d26 : FAILURE in 1h 48m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dd33b94e195b4b648e5053941cc563a2 : SUCCESS in 1h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0bf97ec997af4479afb9b72ce95ef57c : SUCCESS in 1h 27m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/32d611f6c9d74f73a006ca789e378519 : SUCCESS in 1h 37m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bdff33c95d1d410cab37de94ac31f4ba : FAILURE in 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5e3dcace6633f433f5e222bb3be2025fd5a00414","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-24 23:21:02.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"81c0c028710aef7140db12127e4205221f7d67d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-26 00:15:25.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"df582509fc20eb866c523ed49ef293fc954be1f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-26 01:38:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"943d29d5ed527b2f7d72a0ed0a0994d0e4b0b538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 04:03:54.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/02bef10386754aeebb91556d2658713c : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/298079546edd4cb78e70b1e57ea29798 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba780d2ab583452e80c6ac4e56ef79dd : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b574f118acdd44b2ac598bfbe4852372 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a181edb9c8d448392597f833283c8a4 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5ad64db86a5478d8e8a11aff2554164 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe9aa605f49e45a5b78073f8b82d6467 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651f3472ef294da98052d51412c4c98d : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74bf6dfcb77d4319853149214ccea261 : SUCCESS in 7m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e0b7b820732d40bba61b641e90504744 : SUCCESS in 6m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2036f7a06a03422992b4ba19a5203212 : SUCCESS in 11m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/527c62e94dd34075a9a47d5bb3074b04 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9ddcf3e2ffa4f369893c9318ddd2d2b : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c29d7bca3d42473f8b9aa6eee280db1f : SUCCESS in 1h 19m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ee8770d40da49aeb0e9551087861e92 : SUCCESS in 1h 48m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e60cc1f46394a2191813a0234dbbcf1 : SUCCESS in 42m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dc683e615bfe4a1ba3536c7d3a8dfa61 : SUCCESS in 48m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c43e161f2b3c47668bdf293458c72748 : SUCCESS in 48m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a02593aaf7146dca0b553585b9c5916 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ca4a407ecb34435bd5814e0222b08f6 : SUCCESS in 1h 44m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0454094448e459bb89cf764b9ff5fe0 : FAILURE in 2h 21m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8db0224b644940a7b1138f5c56474c0d : SUCCESS in 1h 52m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a4c610fb8044b1fad19faa278ead30b : SUCCESS in 1h 38m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/089fc3cca63a43ca98ab8b2b8ceccfe2 : SUCCESS in 2h 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9b59514d6a4b45caa3eef033b378a492 : FAILURE in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c8073280209549541fd320106edcf6b4f87a730a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-02-27 19:36:08.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"af3af457d41ec65c8a753725f80f319b2323d1ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 22:07:01.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8de2b930102c428e84ea5d87c7ec5e08 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c3e011d05194293ae323cbefde34ec7 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7909831405f2498f95bd880aae26dbb2 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ba9eb2342d4aed8f3f75c5f3610929 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b54725b14c440f85bd076562b569b6 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d81b2e1d61e41018462d7c9b1c59c6a : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5861606ca784e038242c0acf3ba238b : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c21ee7646748339427f1a3fd7be559 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c527b367eb4989bc84482b79a97ac3 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ec639ce618045ec9b074ed128b2d856 : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49f510af2b3449b7b7800724a5ed2905 : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6cab0b1dcc874e09b6b59ec3a9a9d434 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35c54e6ba1614cfdb2bf6764c81862e5 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/391fdccf7c2a47d5b548de6f8536b3d4 : SUCCESS in 1h 05m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e21631c64801494694e812fa537bede8 : SUCCESS in 1h 19m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2579d0239044260a3278e676da719a2 : SUCCESS in 38m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/30c1338f4ec8429e900a84bfa1698d69 : SUCCESS in 48m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c140e3a72764a9f98e082677aae9c2e : SUCCESS in 39m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/976a2aa02cad43c89c854e2191bf936e : SUCCESS in 1h 14m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2183c03ffe4a432f896863e772022296 : SUCCESS in 1h 45m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a579e696763545d0af9994cbd9a20996 : FAILURE in 2h 28m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1d31e4283b44c20a1fdbe36ca7cabc4 : SUCCESS in 1h 52m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d777a0775fa4c95809dc68e54e6bd17 : SUCCESS in 1h 55m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/68191501868c4d54aa05edbfd0dcfb72 : SUCCESS in 1h 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ba534521413c4dc1bd183a27fc79766c : FAILURE in 42m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1ab6510b2b4db48fedaf18f693c01e7a22aa3b05","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-02 19:56:52.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"fe1738359747e752fca3f4761c907e0b915e92bf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-08 16:33:02.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"d9cacd4a1224ca42055b89688e99b697192b271f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-08 19:33:42.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"20ff42c7e043ee388359b06da719f5b27bdeb412","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-08 19:37:24.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"504f0f3a1a00f91684de0fd1fb779c625d75aa7a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-08 19:51:02.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"e803c1d7791b1dbd984e0fa21e2c5763be11705c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-08 20:29:34.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"dbe2432ea83163b0e8b9ebae8c566260e1fa678e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:40:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/364d0b00c7f3414eb48ee3de4b3ac06e : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05711218f575496d93eaa1749d6d134d : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af5d8d50b4044cd0a25f361b79a87b5d : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e430a28d67204cc49e457a6fc4e49f12 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5277a66b3854c998802408e83056118 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98e3163768764e3dac2fadf1d74757b7 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b85981b4f894ddabf4ba406c0692ae5 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6de1a923834cbe8a4039fddca7b6bb : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c277b1192b438f964ce5064dbf364e : SUCCESS in 6m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2edb738837634d1ea923c2cf4f53df1f : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/204a4c81d65942d5b20fc85dd6198d98 : SUCCESS in 10m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/736e604a583e4dc990f2e5957bd16656 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e30ef7c572ab4beca5cd1dcc1e4c1255 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eaebd9b9b30a462f96aa2b59108b8daa : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df6dc85a03344aeb9f4d8a9536bfd987 : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eca09ea568f74c4da49f437301fb0dc8 : SUCCESS in 37m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/250c609617e0494f9a994c2ae15c807d : SUCCESS in 29m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba269e84be6d4f6cb6ef5f6c9ef9a3c3 : SUCCESS in 38m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5faac832aa34640b6615669f9cd4e6c : SUCCESS in 50m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d8f598f1839043c9a1dc2b8e7656e252 : FAILURE in 1h 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d0560b70748e4047974e8e0ffd78c628 : FAILURE in 2h 00m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/147b7eb451ff4d0caa0ddcd42bcc20cc : SUCCESS in 1h 19m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6121facbf93f46408c2aa616e30d6f13 : SUCCESS in 1h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f0d762f07534a0ea781d246c607683b : SUCCESS in 1h 26m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1a5f229c8bf84bfb93c12ea019174062 : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"0b9fd65a0127a52bf1abced0dcee7f694e6595f1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-10 00:50:42.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"09dd1264814311a2cfa18237e7bb0ba44e2ed5c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 00:50:56.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"785bf4e0e23b9cb5aa594426e170fb3ca0b25195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:55:07.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/191b73fe60d14717b74601bc7b6fbae4 : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f92c279d13284cd1a63016e7aeb6c467 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29e04941b9c4b70807bd93aef0197ac : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49395a59c9bf432d958fff5285aee55f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f2b2d13ca454503a95e5a519fa088dd : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4f48473b774d38bd8b36ee8bf34bf5 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c61f23e690c41b197ca89e6a736ed81 : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/615313faa5ba4510b7a928861104b0a4 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a041c0fbd5342f59fd9475b46fbf87e : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/711a6354113644d092ec554f005f7295 : SUCCESS in 1h 20m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e090376962024145a004a8e86a3ecb22 : SUCCESS in 2h 01m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1fb557305c542fea37b9e6a993f1b69 : SUCCESS in 36m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92359ec4e6164fe49fc9b85bc568fe12 : SUCCESS in 43m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30b179af5a61406fbf7826735820bf1f : SUCCESS in 40m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1a08e31e99347918f1b8fb4237ed1c0 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":19},{"id":"ece3a9bb7fe98d2e099bf9cd2daf268855e26583","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"68a8d6799e57479b19ee1867577aa1f72d760b62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:56:25.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd36881caff14234bc59dd51d3a1e817 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e767bb5b6e10441d98795529539595f8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"a5f142c566e5f5654b341c47c23a35b47e8983d1","revisions":{"5ee4fe7d635ef32411f5f3d08e5c2140baea79e5":{"kind":"REWORK","_number":1,"created":"2021-02-17 02:15:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/1"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:15:21.000000000","tz":0},"subject":"Add support for Keystone default roles","message":"Add support for Keystone default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ee4fe7d635ef32411f5f3d08e5c2140baea79e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ee4fe7d635ef32411f5f3d08e5c2140baea79e5"}]},"branch":"refs/heads/master"},"933dc560b5e941ba816b8819609695f9b0e566b9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-17 02:16:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/2"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:15:50.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/933dc560b5e941ba816b8819609695f9b0e566b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/933dc560b5e941ba816b8819609695f9b0e566b9"}]},"branch":"refs/heads/master"},"5413dbdea15f8a3908d98b95a72705dc054e1413":{"kind":"REWORK","_number":3,"created":"2021-02-17 02:33:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/3"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:33:41.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5413dbdea15f8a3908d98b95a72705dc054e1413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5413dbdea15f8a3908d98b95a72705dc054e1413"}]},"branch":"refs/heads/master"},"61d5af7374169a4a1ccd860bae94c3fcb16b53e3":{"kind":"REWORK","_number":4,"created":"2021-02-18 00:23:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/4"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-18 00:23:54.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61d5af7374169a4a1ccd860bae94c3fcb16b53e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61d5af7374169a4a1ccd860bae94c3fcb16b53e3"}]},"branch":"refs/heads/master"},"3c0313613435094715426e5a63561848014ca5c3":{"kind":"REWORK","_number":5,"created":"2021-02-18 23:17:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/5"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-18 23:17:06.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c0313613435094715426e5a63561848014ca5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c0313613435094715426e5a63561848014ca5c3"}]},"branch":"refs/heads/master"},"a8e8c8b23f3ab46358c9289b0f0bc4f170b904ad":{"kind":"REWORK","_number":6,"created":"2021-02-18 23:32:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/6"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-18 23:31:59.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e8c8b23f3ab46358c9289b0f0bc4f170b904ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a8e8c8b23f3ab46358c9289b0f0bc4f170b904ad"}]},"branch":"refs/heads/master"},"a43c9724f5e5bcd9a6350e90e090743e80cc88b0":{"kind":"REWORK","_number":7,"created":"2021-02-18 23:39:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/7"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-18 23:39:38.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a43c9724f5e5bcd9a6350e90e090743e80cc88b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a43c9724f5e5bcd9a6350e90e090743e80cc88b0"}]},"branch":"refs/heads/master"},"1dee19427374be093648775a2be3ad6e2ac13cb3":{"kind":"REWORK","_number":8,"created":"2021-02-18 23:49:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/8"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-18 23:49:44.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dee19427374be093648775a2be3ad6e2ac13cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dee19427374be093648775a2be3ad6e2ac13cb3"}]},"branch":"refs/heads/master"},"1e9c3819fed893d53b9cf2d6cd05a3be46fe7f91":{"kind":"REWORK","_number":9,"created":"2021-02-19 00:07:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/9"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-19 00:07:18.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e9c3819fed893d53b9cf2d6cd05a3be46fe7f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e9c3819fed893d53b9cf2d6cd05a3be46fe7f91"}]},"branch":"refs/heads/master"},"29e7fe86a94653f03dabc97d1836897381c92e44":{"kind":"REWORK","_number":10,"created":"2021-02-19 00:09:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/10"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-19 00:09:08.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29e7fe86a94653f03dabc97d1836897381c92e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29e7fe86a94653f03dabc97d1836897381c92e44"}]},"branch":"refs/heads/master"},"a6594bc4f46718a4de69051047587485a50919fb":{"kind":"REWORK","_number":11,"created":"2021-02-19 00:10:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/11"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-19 00:10:29.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6594bc4f46718a4de69051047587485a50919fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6594bc4f46718a4de69051047587485a50919fb"}]},"branch":"refs/heads/master"},"6a5be37c36df3f8f088687b50478dbfb74eaa69f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-02-19 15:55:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/12"}}},"commit":{"parents":[{"commit":"ce2e0d74258bec3a5ed4983462acee234c67864b","subject":"Add test coverage for SCTP health checker script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce2e0d74258bec3a5ed4983462acee234c67864b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-19 15:55:14.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a5be37c36df3f8f088687b50478dbfb74eaa69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a5be37c36df3f8f088687b50478dbfb74eaa69f"}]},"branch":"refs/heads/master","description":"Rebase"},"19825a5a2441ffbd9906e9ad9bc2a613ec7eee93":{"kind":"REWORK","_number":13,"created":"2021-02-19 19:02:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/13"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-19 19:02:00.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19825a5a2441ffbd9906e9ad9bc2a613ec7eee93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19825a5a2441ffbd9906e9ad9bc2a613ec7eee93"}]},"branch":"refs/heads/master"},"45dba4eee314894beea8c30468398ad4c7c10510":{"kind":"REWORK","_number":14,"created":"2021-02-23 23:25:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/14"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-23 23:25:34.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45dba4eee314894beea8c30468398ad4c7c10510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45dba4eee314894beea8c30468398ad4c7c10510"}]},"branch":"refs/heads/master"},"42dc1e11e100b8505bc1b32075c63c736f31b5c9":{"kind":"REWORK","_number":15,"created":"2021-02-24 15:38:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/15"}}},"commit":{"parents":[{"commit":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","subject":"Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-24 15:38:16.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42dc1e11e100b8505bc1b32075c63c736f31b5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42dc1e11e100b8505bc1b32075c63c736f31b5c9"}]},"branch":"refs/heads/master"},"dfe73cb5787ec27e6cc08e0b0012dc393a622956":{"kind":"REWORK","_number":16,"created":"2021-02-26 00:15:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/16"}}},"commit":{"parents":[{"commit":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","subject":"Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-26 00:15:19.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe73cb5787ec27e6cc08e0b0012dc393a622956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfe73cb5787ec27e6cc08e0b0012dc393a622956"}]},"branch":"refs/heads/master"},"f885e2eef92ebe61690d69a3d6ab7aceda945619":{"kind":"REWORK","_number":17,"created":"2021-02-26 01:38:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/17"}}},"commit":{"parents":[{"commit":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","subject":"Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-26 01:38:39.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f885e2eef92ebe61690d69a3d6ab7aceda945619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f885e2eef92ebe61690d69a3d6ab7aceda945619"}]},"branch":"refs/heads/master"},"d4a2e5c8f0d866cb384414f91cde3098f03a3419":{"kind":"REWORK","_number":18,"created":"2021-02-27 19:36:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/18"}}},"commit":{"parents":[{"commit":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","subject":"Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-27 19:36:02.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4a2e5c8f0d866cb384414f91cde3098f03a3419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4a2e5c8f0d866cb384414f91cde3098f03a3419"}]},"branch":"refs/heads/master"},"a5f142c566e5f5654b341c47c23a35b47e8983d1":{"kind":"REWORK","_number":19,"created":"2021-03-08 19:33:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/775957/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/775957/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/775957/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/775957/19"}}},"commit":{"parents":[{"commit":"c37738ba6f44686d88d9da6d1010e1b8e6af2dc5","subject":"Fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c37738ba6f44686d88d9da6d1010e1b8e6af2dc5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-02-17 02:10:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-08 19:33:35.000000000","tz":0},"subject":"Add support for scoped tokens and default roles","message":"Add support for scoped tokens and default roles\n\nThis patch is the base patch to enable support for Keystone\nscoped tokens[1] and default roles[2] in the Octavia API.\n\nIt strives to maintain backward compatibility and support for\nOctavia Advanced RBAC roles.\n\n[1] https://docs.openstack.org/keystone/latest/admin/tokens-overview.html#authorization-scopes\n[2] https://docs.openstack.org/keystone/latest/admin/service-api-protection.html\n\nChange-Id: I4443d4531dc97d14f8277024baa11ab43e87fb39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f142c566e5f5654b341c47c23a35b47e8983d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5f142c566e5f5654b341c47c23a35b47e8983d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
