)]}'
{"id":"openstack%2Foctavia~776467","triplet_id":"openstack%2Foctavia~master~Icb1430a3709566796e39fbfa962c2541c504b9ae","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Icb1430a3709566796e39fbfa962c2541c504b9ae","subject":"Fix health manager messages about failed failovers","status":"MERGED","created":"2021-02-18 14:23:00.000000000","updated":"2021-03-05 13:30:35.000000000","submitted":"2021-03-05 13:24:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776467","meta_rev_id":"fa7160041ec9a52edea6c45eafdf96e2ab1afda5","_number":776467,"virtual_id_number":776467,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-20 07:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-05 11:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-02-19 18:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-05 11:09:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 18:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 21:10:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-02-20 07:40:12.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-05 11:09:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"5d6d5321d27ef9f01c9032d2e9a65855785b981d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-18 14:23:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6f3fd446d5053fea478731494c05a2c8668e804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-18 15:06:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85604ac00e3e623e69c2cfffe26c9b4638e87fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:00:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b944eb02590476c8c61b2a689bdb660 : FAILURE in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ee27641820848138e0430686e45cc22 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92572f98a4f94a2182a0bf3e30fb8a83 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd96f453e42f48df854a3014d8b209fa : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23307530af6d426fb92c1bf539e4ee93 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1e2f17b9594a5ca5cce332d4b6b334 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4a381f118a41f3be2474a253d8226a : FAILURE in 7m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f677ebb232ee48faaf2bc3a64f4c71be : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5fad99c5c4244addb95e33a582684b19 : SUCCESS in 8m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5520c0ea5194bce8663ddc7a06555b5 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b81877c6fee2451b889d2f62650ced35 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c2c2b2881ea4adabb747cb3881831cb : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b117f54ae1940878a0b07311ef5ed1c : SUCCESS in 1h 55m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4ecaf1311b6347eb8b0be98423cdbb73 : SUCCESS in 37m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7a86a8c910844cfc9e713a66f4eca150 : SUCCESS in 48m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b904bb5f5e974e979fcc899cf78c6897 : POST_FAILURE in 1h 29m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a358db12fe3345088cd7056a29ad4d8f : SUCCESS in 1h 04m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55f0cda99b3c494db0aa637da79b2bbb : SUCCESS in 1h 38m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90d448fed5ba44808b0a2ce802de7357 : FAILURE in 1h 52m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53926bde650841e8bbc39b5c64efd3f7 : SUCCESS in 1h 37m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7b5fd45319b9493cabd4f7b547f5739f : SUCCESS in 2h 36m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/67711c8a73784e63be3e76e0e7a6e42a : SUCCESS in 1h 39m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ef1527e9dbc24795aad5d4a253e6bcd8 : FAILURE in 31m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/install/install-ubuntu.rst","accounts_in_message":[],"_revision_number":2},{"id":"53a47c4086d030dc4273f013ab32a9f1fbdfeac7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-19 18:09:18.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck\n\ncover and docs fix merged","accounts_in_message":[],"_revision_number":2},{"id":"f1cf4a722f3b7c98baa62758d2a78e2f132cc0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:58:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f895871522c744e4bf1e0fab82f01f31 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cbb2b79982246a5a269bd589ce2fa95 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524826eaa30240a59d3fdea6322cb636 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33fa77d52df540a98cd6dc61a6b6592c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c3546e7aa9740cbb83c696eb0664850 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/909fa62adcbf4c7a9d4d4675429e81fb : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d7cc26cb594b56bfb31f203d830b11 : SUCCESS in 10m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/94b7a4c698f44bd6b7311a566b5cc1fb : SUCCESS in 6m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3ca1802e2ba4b3abf5258a1022b94b8 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fc0fa3896f5f4db0b7a56ab6fe76f039 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/737a88a0b3244fb8ae9d0fa592b04cc1 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee7d8dd492da45af97b93e55adc0573b : SUCCESS in 1h 16m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95ddf57ef1674cbbba571262d8b9c783 : SUCCESS in 1h 49m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9e11d947a544578ad5829145ce9e2ec : SUCCESS in 52m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/789b9737bbc743de87874aac81fab934 : SUCCESS in 40m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54b1e411654440978e7100e969859401 : SUCCESS in 33m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c0790bed116409ea1f05f4bdd1553db : SUCCESS in 50m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd4399ae472641f38055b466478df1e3 : SUCCESS in 1h 48m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/605d14d0a1074eee8b7b8edb3d86f3b8 : FAILURE in 2h 02m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f10f3ebf65946c395edeb543c1fa58d : SUCCESS in 1h 47m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0e1023708ceb434dada23b51bcaa356e : SUCCESS in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8af5e3efdf4d4a0fa12871ec639908e3 : SUCCESS in 2h 27m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/89a281b0f119434cb1713370d24b2680 : FAILURE in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b75569936237d0379ce893e211070e457be9fb4c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-20 07:40:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6c9b3cc7f3fbb353825b59580b98e23687977bfe","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-05 11:09:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c05969a216bc5a3794545b2135e3f74d5c8f6db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 11:10:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5a9f03f83703e74eb2603d3af43e4407087566b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:24:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3884e6e77de5428e823e87535c50fa62 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2758f91902c5431f967f3e134a3acc8e : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb2f9e096954a65ad177f788804f55d : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12587a5547c9449e8bf3cb5c583542e4 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180269f57b1c4109a1c555009df5f94a : SUCCESS in 14m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/794c852cc9cf4a2199190f295b26ae20 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f549bb25576c45ae9645493278b97bd4 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f94fbb99ae34042a408c6167e9733cb : SUCCESS in 1h 21m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e95756dd3964106bd319ced9bb0ec07 : SUCCESS in 2h 10m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2025b21ea66648fface75ad2dd03c8a3 : SUCCESS in 38m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4b47f0882e284ff19cd4154909bd1e7f : SUCCESS in 51m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab1a47410bd54bfd9a81c95184d40470 : SUCCESS in 36m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8f6ecd704e8141b781885a97d8330de3 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b45c007462757f96bd63553b02cd73e73e8fb736","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:24:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fa7160041ec9a52edea6c45eafdf96e2ab1afda5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:30:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d9f0ba3d5543319751da0aa1c96cd8 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f6fb8af228b60aebde18dc974d40d6d9cb3dc0a","revisions":{"31efbaedf50b15a409e2aafd6ec341cd3e3578b7":{"kind":"REWORK","_number":1,"created":"2021-02-18 14:23:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/776467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/776467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/776467/1"}}},"commit":{"parents":[{"commit":"38887c6f0f1acde60f0e043c2ba07a636349be6e","subject":"Merge \"Configure rsyslog on Octavia service nodes in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38887c6f0f1acde60f0e043c2ba07a636349be6e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 11:13:50.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 14:22:31.000000000","tz":60},"subject":"Fix health manager messages about failed failovers","message":"Fix health manager messages about failed failovers\n\nThe health manager displayed some incorrect statistics about failed\nfailovers, they were marked as \"successfully completed failovers\".\nThe exception from the failover_amphora function was not reraised when\ncalled from health-manager, now the exception is reraised, allowing the\nhealth-manager to compute correct statistics.\n\nStory 2008636\nTask 41876\n\nChange-Id: Icb1430a3709566796e39fbfa962c2541c504b9ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31efbaedf50b15a409e2aafd6ec341cd3e3578b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/31efbaedf50b15a409e2aafd6ec341cd3e3578b7"}]},"branch":"refs/heads/master"},"3f6fb8af228b60aebde18dc974d40d6d9cb3dc0a":{"kind":"REWORK","_number":2,"created":"2021-02-18 15:06:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/776467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/776467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/776467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/776467/2"}}},"commit":{"parents":[{"commit":"38887c6f0f1acde60f0e043c2ba07a636349be6e","subject":"Merge \"Configure rsyslog on Octavia service nodes in devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38887c6f0f1acde60f0e043c2ba07a636349be6e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 11:13:50.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-18 15:05:53.000000000","tz":60},"subject":"Fix health manager messages about failed failovers","message":"Fix health manager messages about failed failovers\n\nThe health manager displayed some incorrect statistics about failed\nfailovers, they were marked as \"successfully completed failovers\".\nThe exception from the failover_amphora function was not reraised when\ncalled from health-manager, now the exception is reraised, allowing the\nhealth-manager to compute correct statistics.\n\nStory 2008636\nTask 41876\n\nChange-Id: Icb1430a3709566796e39fbfa962c2541c504b9ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f6fb8af228b60aebde18dc974d40d6d9cb3dc0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f6fb8af228b60aebde18dc974d40d6d9cb3dc0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
