)]}'
{"id":"openstack%2Foctavia~776902","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I1236d3c6231a657b2aa53b1e488a4d0fe3215070","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I1236d3c6231a657b2aa53b1e488a4d0fe3215070","subject":"Add validation for VIP network parameters in amphora driver","status":"MERGED","created":"2021-02-22 08:17:16.000000000","updated":"2021-03-04 02:46:00.000000000","submitted":"2021-03-04 02:44:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776902","meta_rev_id":"660a93d379a51daf33ae435f933b72e62bf6e004","_number":776902,"virtual_id_number":776902,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 02:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-01 09:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 00:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-04 00:41:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:17:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-22 09:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 09:48:22.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-03-04 00:41:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"98f507c9d93325e59b26423d675325bc0a62c93e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d995b7d564b6657f4b0004fd18f0fc70a10adc07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:11:55.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/796190ad266d47519ccccc5ffa9a79f4 : FAILURE in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/946bbf6e435d4e01875e1d88f6316383 : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b94825b23cb4d9cb6c124e9df6aca11 : FAILURE in 9m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6384e625bbf4b86a12fb8b2a987ca3e : FAILURE in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f52184893500493183800b8637e5f569 : FAILURE in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f6af0684ba84a81ba860eb89aaa69ad : FAILURE in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1248f9fbc98642e0949b190c7719c384 : SUCCESS in 17m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/283a100f866f4e0698f4f5d9a8811e79 : SUCCESS in 5m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/074b16722ba741b0909a45d4a9637c90 : FAILURE in 9m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b67840ca61ef4c8c83d6dbe3d60bd07b : FAILURE in 11m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c2ffe0ff18014a21975d0374819c7417 : FAILURE in 10m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b20a5178cb3f45f79058e139f45588e3 : FAILURE in 9m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c979fa51c104f5183fe7a8eb882e2b3 : FAILURE in 36m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/15f8344dc9f349a1bf5169dcf68a6165 : FAILURE in 29m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/220ec510fcaa4c79b123e965e942a789 : FAILURE in 37m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a96d993688164734ac7c365f06a87161 : FAILURE in 39m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/23941453902944d4bb6fa9c4baebc69a : FAILURE in 9m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60bc1bcfe4d0476f9afae06fcb165bff : FAILURE in 46m 47s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a72c7e84e4774080a1e3ff2c7836af1e : FAILURE in 45m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c949627f8454aa2baa76d02a2ce6637 : FAILURE in 35m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8aaf7c7b5e45489ab5fe660e5a2b7eeb : FAILURE in 38m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86c457fbed3a41bfbce6da27a09b6c3a : FAILURE in 36m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e18b2900abcb46ec8e1cb9b416f73732 : FAILURE in 31m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c09c6fedb71d4e469ad6ea9712d55e8f : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c7be4577db12fa6d7cd0be8c1de6d24c1be6cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-23 08:21:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9d0b6c8b5b693534a5cbe7ac23f3aeb4fbca57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 10:56:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9da33ef5a7aa428b95239a890e79ef84 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dc10756e3e24757ae462769d01816d4 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/893908e1fa364b08b34227fbf32d4f5f : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c78522d5a19e45889db1cbe8e215bda6 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6a496a5916446e9ad0096fffcfe3d20 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f86d00b78b9462e9e8ee38ebf59c58a : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89b245bc3824520b5d8ee42d90b6907 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3097b57a525c4ea3a3358453536279e2 : SUCCESS in 5m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c1673d4d725e4982a7e69e6568f073b7 : SUCCESS in 8m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3b60370a139a4991a5d4c3f481a85bb8 : SUCCESS in 10m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9dcf5f7224b240c18b87b3c52d6b8f37 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/067e201c83ef45179bf134e72b3daa48 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f41d2ea0a2db41b489f5237dc29a77bb : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/481e99dfd3c84c49b240eb247ec72a75 : SUCCESS in 1h 38m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/978f6d2040ef4bbf9f1a673ef250769b : SUCCESS in 2h 04m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77c755bca320492d8bfc11d9188ee434 : SUCCESS in 1h 45m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4eaeb5bc8811415da63d8910f540d248 : FAILURE in 15m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4625bfe71463488aac2cf7338e3d8170 : FAILURE in 58m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/41c32fe56d9a4b8a9586185e4d5fe468 : FAILURE in 41m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6df83854fba949db9e1ebaa57dd341c9 : SUCCESS in 58m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1cd69d3cb6354a228921e44c4421cd14 : SUCCESS in 2h 19m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb4c4f7317624cb1842447327522684f : SUCCESS in 41m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/abc7939c62ce43668a0be3edfdfcc639 : SUCCESS in 40m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/214956f454e34e06a8cceb82cfb41d07 : FAILURE in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"570155384bac6ad08d9b9be1b6902aa23e3312ee","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:48:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b23912b1a463ecc04046779502472cbbdb0b2b2d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:41:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"6e16a591f3099088421113a05d79c236ba5ef3e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:41:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16bf03dfb8e3089a681628cb3520596b4011ba38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:44:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b62cc8bf1a460890283865b67c4288 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7eca2c46af465f9508e6dfec197bf3 : SUCCESS in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/820f1c56cbaf4692b336bd567ce61be7 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae7ad6ecfea7452da1706dfaa38960d1 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbf4c946da12460eaeac1bcd3fbfcf25 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6f0a9d818d493ba061e35c95f32b42 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cbf490581e64fbdbd61d87a6a23a52e : SUCCESS in 4m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3d32df218d8a4381bde5e3e3fdc8257f : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc54df57f2674df986dcbf918e40a9c6 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9cf4c4deec4e4422b4ef25664edd7afd : SUCCESS in 1h 20m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/25364ace0f5a460bb9af5e0cd8102787 : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae5b4c9102524c8fba42000f9ecccdfe : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":2},{"id":"934ddc4ccd10adbf5791baee16449dae161f5232","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:44:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"660a93d379a51daf33ae435f933b72e62bf6e004","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:46:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b527786f13405d9ac6f12c6ba10e07 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d65c4906828466cbd45cfc08f5d8f59 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6767a88b61121ac454d8b939ab15680473087e87","revisions":{"5d68db197f1145a1f2f28390d78545654edadcbe":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:17:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/02/776902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/776902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/776902/1"}}},"commit":{"parents":[{"commit":"131c69608dcb3c621f2dde12d4c4b0ae1c87bf45","subject":"Fix amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/131c69608dcb3c621f2dde12d4c4b0ae1c87bf45"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-12-14 17:24:10.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:17:05.000000000","tz":60},"subject":"Add validation for VIP network parameters in amphora driver","message":"Add validation for VIP network parameters in amphora driver\n\nSome network parameters can be validated in the API, it would avoid to\nhandle exceptions in the worker when plugging networking resources.\nThis commit validates that port_security_enabled is True on the VIP\nnetwork when using the amphora driver.\n\nStory: 2008449\nTask: 41422\n\nChange-Id: I1236d3c6231a657b2aa53b1e488a4d0fe3215070\n(cherry picked from commit dda1d8665c1b90e3494b22bc0de1853a43b446a6)\n(cherry picked from commit 250302bf0359758ffe59c10259db04fa550d58ff)\n(cherry picked from commit a0cb2a0df516e2d80d33693c62d875ccf60bee10)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d68db197f1145a1f2f28390d78545654edadcbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d68db197f1145a1f2f28390d78545654edadcbe"}]},"branch":"refs/heads/stable/train"},"6767a88b61121ac454d8b939ab15680473087e87":{"kind":"REWORK","_number":2,"created":"2021-02-23 08:21:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/776902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/776902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/776902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/776902/2"}}},"commit":{"parents":[{"commit":"131c69608dcb3c621f2dde12d4c4b0ae1c87bf45","subject":"Fix amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/131c69608dcb3c621f2dde12d4c4b0ae1c87bf45"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-12-14 17:24:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-23 08:20:31.000000000","tz":60},"subject":"Add validation for VIP network parameters in amphora driver","message":"Add validation for VIP network parameters in amphora driver\n\nSome network parameters can be validated in the API, it would avoid to\nhandle exceptions in the worker when plugging networking resources.\nThis commit validates that port_security_enabled is True on the VIP\nnetwork when using the amphora driver.\n\nStory: 2008449\nTask: 41422\n\nChange-Id: I1236d3c6231a657b2aa53b1e488a4d0fe3215070\n(cherry picked from commit dda1d8665c1b90e3494b22bc0de1853a43b446a6)\n(cherry picked from commit 250302bf0359758ffe59c10259db04fa550d58ff)\n(cherry picked from commit a0cb2a0df516e2d80d33693c62d875ccf60bee10)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6767a88b61121ac454d8b939ab15680473087e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6767a88b61121ac454d8b939ab15680473087e87"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
