)]}'
{"id":"openstack%2Foctavia~776903","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I6dd8e99b07ff557674871cb503dece96a9df3ada","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I6dd8e99b07ff557674871cb503dece96a9df3ada","subject":"Disable auto configuration of new interfaces in NetworkManager","status":"MERGED","created":"2021-02-22 08:18:22.000000000","updated":"2021-03-05 21:40:03.000000000","submitted":"2021-03-05 21:38:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776903","meta_rev_id":"ba604184eeb503be86bd1acabbda36edafd841d8","_number":776903,"virtual_id_number":776903,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 21:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-03 22:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:18:22.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-22 10:37:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 22:29:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 09:25:38.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"37662fe4b65c845b1422bbd7dd2cb2a7969dafc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:18:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3a71fb5f23174f579c2ad7c856ae34c3dd5efe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:37:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc407aa74168449d8e112c1ed665f018 : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d78a675d650442aa97e4b9c736270e1 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c3c62632834a189fdfa86dc9146dea : FAILURE in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45dee879b26d4e34b7f12eb96450665a : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9cd48dec07c45deb36549a236e9b139 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c936ef7538df4b68b4edb8649cd2d882 : SUCCESS in 17m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2dcbd7b1dd0c403da4ecb59419d8db68 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e3dc051188d44980a2b7348a5a1ead63 : SUCCESS in 10m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1bb970f4f28d43a2aa17e3453837ec7c : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a920f52f853c4d7abac575b50b10799a : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a382688550f44c179f1bd3378acb7708 : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed82222c78c14ec5ae0efcf700a1f090 : SUCCESS in 2h 13m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36b702178a0c40cc92aa52eb6c2aceaa : FAILURE in 1h 21m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a6b64563e7824b83a377e0e3fb508775 : SUCCESS in 40m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9c061ab407b4c79964193cfef511274 : SUCCESS in 38m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea32c928747c4a4a9d910b85c59f0eb7 : FAILURE in 32m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3a1b79c6c204d9ea083bed38943b4f8 : SUCCESS in 2h 05m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f899c43abcd5457b9caa9264460f9119 : FAILURE in 2h 01m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce6a3d11deb14e20bfd09aef23b552e5 : SUCCESS in 2h 01m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c861269877da4b2aab89ca170dbd85d0 : FAILURE in 1h 55m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/53ffcafa208f454192b982deb84385a4 : SUCCESS in 2h 07m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6eabe1127c734604bbb5e64c5e51b462 : FAILURE in 1h 49m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/common/policy.py","accounts_in_message":[],"_revision_number":1},{"id":"343e99b3423e9d1a1280ff202cf7b91ab6f5a186","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 22:29:48.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\nLTGM","accounts_in_message":[],"_revision_number":1},{"id":"5dec62d6e4be97cb8ee97d2919f0a14aec7573c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a7a24f6f1447e08acc5896c1a1cb11 : SUCCESS in 11m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19f48c780dc143aeae9ac7fe715e2e41 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2cb228f3244aa3a908bd410e4ed88f : FAILURE in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bbad8b5529440a3a728695313b4c0b5 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da0d0456415d489f85953f8a7634967c : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94550c9a0519469ca52c10ac4f043ad5 : SUCCESS in 16m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5a6c7935f3b041508c36dc4ae1f56736 : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b13671f93ca548e68fb8e4b9f8d15b7f : SUCCESS in 10m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/857927999cc94513951f41d0e352a024 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e4305e3ebd54a6babe7a2bfad647065 : SUCCESS in 10m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca39c25dda744d50b9f6dcec85196781 : SUCCESS in 1h 46m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f7c5e7063b741f8a805aa883cd3ee86 : SUCCESS in 1h 35m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90f1922ec759457eb169dd764e4bbe73 : SUCCESS in 38m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b07720f42fe543d1a70f107871c6b254 : SUCCESS in 37m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caa8d0ac6a114ebf9c64ab3f825edfe3 : SUCCESS in 46m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19ccf4b7628542f3bbc135fbde79a8a9 : FAILURE in 37m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/15139ccf666b44bfbaf67e8488e22eec : SUCCESS in 1h 50m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/931f03e60fd74fdeabe211a96b75307a : FAILURE in 2h 10m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e000712de2f04e228b01f35d9715db2e : POST_FAILURE in 1h 26m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd5ba2ffc7664a7a8bf1294164d7dfd1 : SUCCESS in 2h 00m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8570a18f5ff54091b190ea80b64413d5 : SUCCESS in 2h 05m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/51310f2446004c108b14d93bd42d5e7d : SUCCESS in 1h 24m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/common/utils.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/controller/worker/task_utils.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py","accounts_in_message":[],"_revision_number":1},{"id":"04f2ecce19e32acd9152975617701f3e4d5ef521","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:25:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75ca7c0bfe2085f1ab87c3b299b4733e1e3218f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:00:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39ede05c22d448a28fa6fcf3781ba33a : POST_FAILURE in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f29f1283a549f980db15cf5c4c171b : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c1d50fff64f4d71a54db294f4a8bdaf : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d8dff8515714659a9d66bc4b66c13a8 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9567f639a0364c6fab8d564459652ad5 : SUCCESS in 12m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6af925a2fe8747bcbf6b981e4cae02b9 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d63cf4ddd64a4ee8a8845fc0536ace27 : SUCCESS in 7m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2b46c4e5f47c4ba2b5c47ab29b842852 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f04cb6c792cc48bcbf29c27232d7bf4b : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/185571a90e2940c39ab2d1268d96e199 : SUCCESS in 1h 19m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b69949838794d66a34bf6e7ac4e7089 : SUCCESS in 2h 05m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/177e6f4a3ea744d3adc407f61d48f866 : SUCCESS in 44m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6f9f84c0707947e0bcafd7d24d0a7173 : SUCCESS in 39m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23b4110d4c2949d2a83fce2497dff044 : SUCCESS in 48m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63e562950ebb4da080146f838374ed88 : SUCCESS in 1h 05m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/269f8927fe664d078a7b8f4b84d4724f : SUCCESS in 1h 59m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0372cf0452904269b61679c828efa9ca : TIMED_OUT in 2h 32m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de4d33708be34d27aa8ac522e4d246d6 : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e32e2830989341fb9259c8a54a11ef43 : SUCCESS in 1h 52m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e9176363b00940bc8bb57b8baea1be35 : SUCCESS in 2h 16m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2724a99771584b55adf775246ca7d2f7 : SUCCESS in 2h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07d761444b89380cfb1eb2ae2c327c61183f1d1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-05 16:34:32.000000000","message":"Patch Set 1:\n\nrecheck\nPOST_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2efd5a592fd8883fa5ff0f03e950c8422d9e4b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ba7f002344d4539aaae9d37388f0eee : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5258462ff484526928adf39db90d332 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1596b9016cad44f38df1e3e22841a41e : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1787fd93be1a454c8b9de9f7f8f4296d : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e438f0120e46a4a1cb1d7415e6632d : SUCCESS in 12m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/29a5266192ce4dafa05d6299f10c3b50 : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/154b8468c69b4a6390529cd502298739 : SUCCESS in 11m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/309fd944e19e4e9b802cccd25888c5f6 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/123fbd60f6e34ac9b7df38dc47598451 : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6504c5de467947bea238db52071078c6 : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1e68c4f22ee41a38532b08939cdd81e : SUCCESS in 1h 55m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2fd16f793daf423493de49a519c26151 : SUCCESS in 42m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95ccc87ac0fb419cb0937c41fc57b7db : SUCCESS in 43m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/753a3ce92dd145479125c1560354c815 : SUCCESS in 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3d519b1d4634ff099526af0c73166e7 : SUCCESS in 52m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8cae314da73f4dd3a85b94788ae78c0e : SUCCESS in 1h 41m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8fb272aeaac94bf38ea1de69ef389f96 : FAILURE in 2h 12m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c510d5a3b77a4b91a1ad303fbf6c9d40 : FAILURE in 1h 54m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5768a1391c5a486bb3b5c129994f076c : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b4a3b2b5378b49759af033e5be221d7f : SUCCESS in 1h 40m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2fb70d13d9b64825857c42f132714dda : SUCCESS in 2h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a841d76ab8be84f1bceca6d3a632de0ac97a031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:40:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c27d0b8913dd176dee72d5aef1c057704cc77f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:38:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508ae71d67fe42618c0211b44c96c6ed : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92b27e17d27644f78c312af95261d16d : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac9bc17e2f854e099192b26c6d3b60fe : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92359e2c965449d09454028771d3ff9b : SUCCESS in 17m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a8c1e5aa7404c12ba74eecd77cb2cbc : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf58dcce0e1545b98f8e093188326594 : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/728c2e240e31443ca4bdf10ec3afab6a : SUCCESS in 1h 30m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97daa8f1eb25490abf0751dabd89cec9 : SUCCESS in 1h 54m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cc03ab1ed76340dba0679767b04e49c9 : SUCCESS in 44m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bbfd7cc055a0437c9e885fea99c29bd6 : SUCCESS in 38m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b77663f287a84b74bd9a3d22e1a9add6 : SUCCESS in 1h 07m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aeaba727f2d7414c98473e3a79ec706b : SUCCESS in 1h 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"435a2f6123f866c5b1ce0e9fd3a1cc953aeb705f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:38:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba604184eeb503be86bd1acabbda36edafd841d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd7b90368014af7a2d1c0eda81ecee2 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9f622b335dd4f95f1a6004bb94159f891ca0980","revisions":{"b9f622b335dd4f95f1a6004bb94159f891ca0980":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:18:22.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/03/776903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/776903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/776903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/776903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/776903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/776903/1"}}},"commit":{"parents":[{"commit":"b5c6c884a4a4de48eea7ea08ba865f08fbfcf972","subject":"Merge \"Fix operating status for empty UDP pools\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b5c6c884a4a4de48eea7ea08ba865f08fbfcf972"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-06 21:26:07.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:18:11.000000000","tz":60},"subject":"Disable auto configuration of new interfaces in NetworkManager","message":"Disable auto configuration of new interfaces in NetworkManager\n\nNetworkManager in Centos images configures new network devices as soon\nas they appear in the default namespace, it means that we might have\nconflicts between the management interface\u0027s routes and address and the\nnew VIP or member interfaces\u0027 routes and addresses during a small period\nof time before they are moved to the amphora-haproxy namespace.\n\nNow, the \"no-auto-default\u003d*\" option is enabled in NetworkManager, it\ndisables the configuration of new interfaces, while the management\ninterface is still enabled/configured through cloud-init.\n\nStory 2008599\nTask 41773\n\nChange-Id: I6dd8e99b07ff557674871cb503dece96a9df3ada\n(cherry picked from commit a518cefda1332042d50e5ce2814f0af933c62457)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9f622b335dd4f95f1a6004bb94159f891ca0980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9f622b335dd4f95f1a6004bb94159f891ca0980"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
