)]}'
{"id":"openstack%2Foctavia~776905","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I6dd8e99b07ff557674871cb503dece96a9df3ada","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I6dd8e99b07ff557674871cb503dece96a9df3ada","subject":"Disable auto configuration of new interfaces in NetworkManager","status":"MERGED","created":"2021-02-22 08:19:48.000000000","updated":"2021-03-05 13:21:41.000000000","submitted":"2021-03-05 13:07:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776905","meta_rev_id":"c1f9094d31de639561994d6ba0661497d7714631","_number":776905,"virtual_id_number":776905,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-03 22:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:19:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-02-22 10:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 22:34:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-05 09:34:31.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"c836dd644c46e466eb8ff60412f71a70ae1df6ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bd5010cb90b15a48831f4c52c6f609bb1b4ce68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:40:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49e16f65bb284d3b9dc68a85b2391a5b : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31dec6aeec4f4ba4bcc85bd747e80d6e : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59953346cdac44f284b675d6a65cb80d : SUCCESS in 8m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/009f6f985db04923b1f995a772cfc767 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96464afad8234296bb5e437cb25fc472 : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff99df9ed9af4e15b37d559c83eb9970 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f8f1c731024e4aae1aeca0301f037c : SUCCESS in 14m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/579a29408b8540a282a4709abb984644 : SUCCESS in 11m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c3734b33c5994f30941e1d0f3faa5240 : SUCCESS in 13m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/87151a217a394bf586c8b509195f43c7 : SUCCESS in 10m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c3a4aaacc5a4520a09e0f540c00580c : SUCCESS in 11m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44ea2a42b6b84fa8bb755076b2638e45 : SUCCESS in 1h 23m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0415121b856443018b1c1340d7dad3ed : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96ba1aa56efa4fc2ada13495840a0591 : SUCCESS in 2h 02m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9d9c1abd44834593aa1b04ec48b7d136 : SUCCESS in 2h 04m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b6dc809a438247ceb5a522d572005ba3 : FAILURE in 10m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/785ce8ed574d4dfc940a2e8355118bd2 : FAILURE in 46m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/76c2779f6b534bed8ed9eb08aee3cabe : FAILURE in 53m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5873f66094b4ca39e9692f9131bc186 : SUCCESS in 41m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7922835f6d554fa6ad06a4e3ed51b986 : SUCCESS in 1h 47m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4935c5d4d27f4400aadc1b6efae33ec3 : SUCCESS in 52m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ac20e31cb1284debb00519a3ac7f94bb : SUCCESS in 41m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ef5de022ef3645fbaf2905444a1b52e4 : FAILURE in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e69cf1577a23864e0baa8d3c6e725080c9f027b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 22:34:28.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"3e49607a19fddf6cbc55e1af90669ab0601df531","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:34:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e002b684d1d5cc74b4f3274ca0fd98554ed3ec4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:36:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d88e58045b0b9eb01fb6e18e4223f973211c5dda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ec1ff481e1e4faeb9c409f9ef475cc1 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad7c69e4490f47d2a31d00fca0fdb08b : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4d219b89cab40c6962c46454d527f69 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/023045812ccd4c569e68e1cb44903790 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/519556be1e5d4f98b5a8c43cbb170bb2 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21056e02080d488aaad1ea055770d874 : SUCCESS in 11m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f099731ba704fd985b922bf1695b254 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e913b88625ec4fc9a5debec240a2772b : SUCCESS in 9m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddebfc4c19db47e2a9084c1c18430b0c : SUCCESS in 1h 03m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dc6e059a268f41d1a247fe683104c1bc : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30a09729dd3d42cf824d594a3fc4def9 : SUCCESS in 1h 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1391085a3d0598d9bce7a5fa2f007ab1035bd0dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c1f9094d31de639561994d6ba0661497d7714631","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b073a999d674687a8f61097f6df0fc2 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0288729295df217788a7e2c42034df84188807f8","revisions":{"0288729295df217788a7e2c42034df84188807f8":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:19:48.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/05/776905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/776905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/776905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/776905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/776905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/776905/1"}}},"commit":{"parents":[{"commit":"131c69608dcb3c621f2dde12d4c4b0ae1c87bf45","subject":"Fix amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/131c69608dcb3c621f2dde12d4c4b0ae1c87bf45"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-06 21:26:07.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:19:38.000000000","tz":60},"subject":"Disable auto configuration of new interfaces in NetworkManager","message":"Disable auto configuration of new interfaces in NetworkManager\n\nNetworkManager in Centos images configures new network devices as soon\nas they appear in the default namespace, it means that we might have\nconflicts between the management interface\u0027s routes and address and the\nnew VIP or member interfaces\u0027 routes and addresses during a small period\nof time before they are moved to the amphora-haproxy namespace.\n\nNow, the \"no-auto-default\u003d*\" option is enabled in NetworkManager, it\ndisables the configuration of new interfaces, while the management\ninterface is still enabled/configured through cloud-init.\n\nStory 2008599\nTask 41773\n\nChange-Id: I6dd8e99b07ff557674871cb503dece96a9df3ada\n(cherry picked from commit a518cefda1332042d50e5ce2814f0af933c62457)\n(cherry picked from commit b9f622b335dd4f95f1a6004bb94159f891ca0980)\n(cherry picked from commit 7b2424685e9f29116d48cf2d0922f552fd4c2fdd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0288729295df217788a7e2c42034df84188807f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0288729295df217788a7e2c42034df84188807f8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
