)]}'
{"id":"openstack%2Foctavia~776907","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I736816247131715f5c385b4680614ec3218a2ad7","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I736816247131715f5c385b4680614ec3218a2ad7","subject":"Fix pools going into ERROR when updating the pool","status":"MERGED","created":"2021-02-22 08:21:27.000000000","updated":"2021-03-05 13:16:26.000000000","submitted":"2021-03-05 13:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776907","meta_rev_id":"9e20f35c7a2603291ce15c4e94a46de36cf5acd6","_number":776907,"virtual_id_number":776907,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 13:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-05 09:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-03-04 00:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-03-05 09:26:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 08:21:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-02-22 11:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 09:26:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"7e5d95b4f7e23ebd807def82136d8ff00249bb31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-22 08:21:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90844d3fd0ebc4363be722abb40f9c04f393196c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 11:03:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd1ecf3e04284a7b8d8d481a7558acdd : SUCCESS in 13m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d854612097274edbb5c37f5bc71c7688 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3725de4ea046ed8a70adcb4779c1a7 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7041bb6806604501acb50528b109bd00 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a468eb333cce403a814a1dc095c2e7ac : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be56d014a7bd4d3090b442037c2f1e62 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3bd0ba6955544b3a8ce423e7dd38b6b : SUCCESS in 7m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b1bdffd510c141e8ab341768a9462f81 : SUCCESS in 8m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b76f62a43b7e4056bb60f9c0e1cfd4bd : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0e54456264d41d0a72216de1c45f21e : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/654ff76fd95d4677aad639422ff3b813 : SUCCESS in 1h 17m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcec099be2234a33907a356e318ad1e1 : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a768ebcde6624426aba064869998460a : SUCCESS in 43m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2aced24d2ce744beb951e2a66012711a : SUCCESS in 40m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e710e2ffb7443fdbfb3ee061aca1740 : SUCCESS in 42m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eae047022dc642a69c3f7d1637d91b20 : SUCCESS in 1h 13m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/caa6550ba4184b53adfa936ef8747351 : SUCCESS in 2h 10m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a90e8e646cfd412b9f391710547f454c : FAILURE in 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11956bc1fd444f9aa803a78000493ffc : SUCCESS in 2h 18m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59e716024622a8dd426b48bd05059716888fefde","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:20:18.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"bc8194673770902c3de806aaef485372b3995c00","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-05 09:26:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81cce23fc3f88de6e017efd562ff6678dc107f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 09:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc37208a1a8e7761f45649a89327c56059973195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8540712c7a04f18b2320f249b87f9c4 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902825cd04d14fa8bc41f87ee6e59b5b : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de22f59328e045f980487d9e64b98182 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/816c3f3773564577b23b410f66e504e8 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0964335d39554647a2578a56c34a4bb0 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5f88af051647de99ed51c57814caf0 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0743bbbe94c49fbb2302e5302dfc04d : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c3c5f0e22ca4de8ba2e12b4b79f2dc7 : SUCCESS in 1h 24m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df4129bb76654d829d587da016bbbf25 : SUCCESS in 1h 30m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37f3c62cf40047ea908c715122bdd18c : SUCCESS in 34m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4aa7ac6ff4a142bca0328dff60992b42 : SUCCESS in 35m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11e90e188b5342d082a8d5ef91bd6d05 : SUCCESS in 43m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/624ab22e9072445c98546f5d7c509556 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"96f94e16df7f5b56d58ab3ae23d9a2b49dc6aa25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:07:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e20f35c7a2603291ce15c4e94a46de36cf5acd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 13:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a21987fec3424184db3082f1d3bb18 : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/516d67123956452db4b1136da8220191 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebe234eb252ffd4a051eab6bf05e587430d8432e","revisions":{"ebe234eb252ffd4a051eab6bf05e587430d8432e":{"kind":"REWORK","_number":1,"created":"2021-02-22 08:21:27.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/07/776907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/776907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/776907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/776907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/776907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/776907/1"}}},"commit":{"parents":[{"commit":"6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe","subject":"Merge \"Add missing log line for finishing amp operations\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ef1564dbab34fcd1fef6a0d3bdaeb129e8cf9fe"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2021-02-22 08:21:11.000000000","tz":60},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n(cherry picked from commit 370aa4e61c421311b06d986aafdf123cd6737fd8)\n(cherry picked from commit 24e3a4d93a5c0ae6dcc880b35bd847ba0393e040)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe234eb252ffd4a051eab6bf05e587430d8432e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebe234eb252ffd4a051eab6bf05e587430d8432e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
